)]}'
{"id":"openstack%2Fpuppet-octavia~691780","triplet_id":"openstack%2Fpuppet-octavia~master~Iecc817f44ee34a737be6085ea752a878c8b7095d","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1850236","hashtags":[],"change_id":"Iecc817f44ee34a737be6085ea752a878c8b7095d","subject":"Fix rabbitmq dependency","status":"MERGED","created":"2019-10-29 01:32:16.000000000","updated":"2020-04-03 08:45:07.000000000","submitted":"2019-10-29 01:59:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691780-1572314371918-edb63827","meta_rev_id":"896c6c2eac67707b5d06b68ae256a1036156f0b9","_number":691780,"virtual_id_number":691780,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-29 01:59:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-10-29 01:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-10-29 01:54:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-29 01:59:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2435e02a5a97351017159ba7e61a1247dbdcd941","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:32:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3b24352204c51325cece3dcecb007870283d2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/eb61b18611ea40b399a0b30daaf952b1 : SUCCESS in 3m 52s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6032e0cd71eb4e5cab17eb8afc29643b : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"420e6dba2e4acb601b034a495de6c0a7fdd72a95","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-10-29 01:54:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"625e6f62584e52767ca74b116862f8f66897bd8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:55:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac77d70895d3dd325f16dc87482699af5a2f645f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:59:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/fa3dc91158b544a090ccc97bd22bd14d : SUCCESS in 2m 23s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de7ef6696b214c45b39ae33a3f92b82a : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6cbea94110b746e53f33e9e44aa6fe4bdb00605c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 01:59:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee360192e32d1898da5d34625abcf914a21ca5bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-03 08:45:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e322e9b2dd9505fc1b8fbc58a663a6fef42f386c","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"904c732828c50983b48d6fb6d5a86a20e1fd5927","revisions":{"904c732828c50983b48d6fb6d5a86a20e1fd5927":{"kind":"REWORK","_number":1,"created":"2019-10-29 01:32:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/80/691780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/691780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/691780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/691780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/691780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/691780/1"}}},"commit":{"parents":[{"commit":"8fd7b2e0ff231a602167af98f8fe7729c872b319","subject":"Add keepalivd vrrp parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8fd7b2e0ff231a602167af98f8fe7729c872b319"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-10-29 01:31:05.000000000","tz":480},"subject":"Fix rabbitmq dependency","message":"Fix rabbitmq dependency\n\npuppetlabs/rabbitmq [1] is deprecated, and max version is 5.6.0 which can\nnot be fulfilled by this dependency.\n\nCorrect it to puppet-rabbitmq [2] instead which is being actively maintained\nand has the right version.\n\n[1]: https://forge.puppet.com/puppetlabs/rabbitmq\n[2]: https://forge.puppet.com/puppet/rabbitmq\n\nChange-Id: Iecc817f44ee34a737be6085ea752a878c8b7095d\nCloses-Bug: #1850236\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/904c732828c50983b48d6fb6d5a86a20e1fd5927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/904c732828c50983b48d6fb6d5a86a20e1fd5927"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
