)]}'
{"id":"openstack%2Fpuppet-octavia~699067","triplet_id":"openstack%2Fpuppet-octavia~master~Ibdce9e9570eb0cdf02fd7595eb570e2545d8ceec","project":"openstack/puppet-octavia","branch":"master","topic":"699067","hashtags":[],"change_id":"Ibdce9e9570eb0cdf02fd7595eb570e2545d8ceec","subject":"Revert \"Add parameters for amphora anti-affinity\"","status":"MERGED","created":"2019-12-14 15:25:26.000000000","updated":"2019-12-15 08:04:59.000000000","submitted":"2019-12-15 08:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699067-1576397009305-b807434a","meta_rev_id":"53fc7bb811d0ace03860423cba38869f96a2332b","_number":699067,"virtual_id_number":699067,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2019-12-14 15:31:05.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-15 08:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-12-14 16:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-12-15 07:15:18.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 15:25:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-12-14 15:25:26.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-14 15:25:26.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2019-12-14 15:25:26.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-12-14 15:25:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-14 16:04:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-15 07:15:18.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-12-15 08:03:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc8fa0f3b2d4c76693d794447af99a03ed906150","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:25:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59afd4077fcd608e6faadbcda67df9888c393a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 15:25:38.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"3130d4159a30796a69b09796840b27855c1d687c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:25:55.000000000","message":"Patch Set 1:\n\nThis broke the module and was backported to all versions...\n\n2019-12-13 16:12:27 +0000 Puppet (err): Evaluation Error: Error while evaluating a Resource Statement, Duplicate declaration: Octavia_config[nova/enable_anti_affinity] is already declared at (file: /etc/puppetlabs/code/modules/octavia/manifests/controller.pp, line: 189); cannot redeclare (file: /etc/puppetlabs/code/modules/octavia/manifests/nova.pp, line: 48) (file: /etc/puppetlabs/code/modules/octavia/manifests/nova.pp, line: 48, column: 3)\n\nThe code to handle that config option already exists in octavia::nova class.","accounts_in_message":[],"_revision_number":1},{"id":"0a68bc3e72b4d80797681b7510b501b8fe9ae986","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:27:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9efa6eb22248e39acca5da1359ca2737d51f25d9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:30:45.000000000","message":"Patch Set 2:\n\nThe octavia::nova class was introduced in the Stein cycle and backported all the way back to Queens, so all this functionality already existed.\n\nhttps://review.opendev.org/#/q/I7d271c0184750aa90be59a25944c29fec3c3afd2","accounts_in_message":[],"_revision_number":2},{"id":"088d0308b12c6eae02f4a81806d5c9e18c4d7373","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:31:05.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/#/q/I7d271c0184750aa90be59a25944c29fec3c3afd2","accounts_in_message":[],"_revision_number":2},{"id":"1086a8bffdb1ff363100c4df619233b358d8b7aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 15:51:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ffc05a2791548728e800905fa62eadb : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19f89106d6514dfc991485084af65f09 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ceb29f3ef334406bbc6696eaff4a9eac : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/12b8fcd14e8e4339acd0550e2db5951a : SUCCESS in 12m 20s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/444a845e559841e4b4262d56f01a887b : SUCCESS in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de51bcc5fe324ad2b2fd6d5710c1f964 : SUCCESS in 9m 14s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/709de78196ed4736a0c9aa6febe31ae4 : SUCCESS in 22m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bd8e4f3728b4ab49adf3cd673549c37 : FAILURE in 16m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e471f12292a84dfdbc9063b5fc990b1a : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"69835c3b1638a716bbbe03f9d5a0123f89bd90f2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-14 16:04:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9e983e0cce750dd96ad41a277a47b1b275a83d6d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-15 07:15:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc1efd7985f27696c8c9e2eb846d319e754a1ecc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 07:15:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4b64d10fb7c224383c07a7780cfb42055d046d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:03:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eded7ed84a3d45cb8fa2e0bf1bbaf227 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/194d2a1936c7493d8afed7fa4b84e8fd : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7497911e0974fcc80136faf03635204 : SUCCESS in 3m 53s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a182922f90184ec28463c7dd8f834956 : SUCCESS in 17m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95fe9c1d22c147089d52a098af0565eb : SUCCESS in 12m 11s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4be7ebc4a48e4d2fa5b0e933b840fb3c : SUCCESS in 23m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab5fa2efaa84aadb88eea04a1338b9b : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"a8482e6ef1a5e745df7f39400a11adf861458a29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:03:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48b0793bac8a99b1a0804a4c103685a8d4f94632","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:04:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/656a444d8f104209a6e77bd2072c58b3 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5a5219dd98ba3989575b0c763b5f7460cf3bd45","revisions":{"3bf9737ecd87a0767f2b9b88663f14f83e808256":{"kind":"REWORK","_number":1,"created":"2019-12-14 15:25:26.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/67/699067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/699067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/1"}}},"commit":{"parents":[{"commit":"4bd67a42d0d812553db2caba2b56e01e90beaadb","subject":"Add parameters for amphora anti-affinity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4bd67a42d0d812553db2caba2b56e01e90beaadb"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:25:26.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:25:26.000000000","tz":0},"subject":"Revert \"Add parameters for amphora anti-affinity\"","message":"Revert \"Add parameters for amphora anti-affinity\"\n\nThis reverts commit 4bd67a42d0d812553db2caba2b56e01e90beaadb.\n\nChange-Id: Ibdce9e9570eb0cdf02fd7595eb570e2545d8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3bf9737ecd87a0767f2b9b88663f14f83e808256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3bf9737ecd87a0767f2b9b88663f14f83e808256"}]},"branch":"refs/heads/master"},"e5a5219dd98ba3989575b0c763b5f7460cf3bd45":{"kind":"REWORK","_number":2,"created":"2019-12-14 15:27:24.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/67/699067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/699067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/699067/2"}}},"commit":{"parents":[{"commit":"50228311a3ff1ac3693d8375d5c2e43fdd890579","subject":"Fix stdlib version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/50228311a3ff1ac3693d8375d5c2e43fdd890579"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:25:26.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:27:18.000000000","tz":60},"subject":"Revert \"Add parameters for amphora anti-affinity\"","message":"Revert \"Add parameters for amphora anti-affinity\"\n\nThis reverts commit 4bd67a42d0d812553db2caba2b56e01e90beaadb.\n\nChange-Id: Ibdce9e9570eb0cdf02fd7595eb570e2545d8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
