)]}'
{"id":"openstack%2Fpuppet-octavia~699069","triplet_id":"openstack%2Fpuppet-octavia~stable%2Frocky~I3c8dda3bbf8a09296fbe9ea3b7ae92b46dadf94d","project":"openstack/puppet-octavia","branch":"stable/rocky","hashtags":[],"change_id":"I3c8dda3bbf8a09296fbe9ea3b7ae92b46dadf94d","subject":"Revert \"Add parameters for amphora anti-affinity\"","status":"MERGED","created":"2019-12-14 15:28:17.000000000","updated":"2019-12-15 08:04:47.000000000","submitted":"2019-12-15 08:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699069-1576397007489-70b46864","meta_rev_id":"68c65cd576982beb43c407f3bdac283b91fb5a2a","_number":699069,"virtual_id_number":699069,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-15 08:03:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-12-14 16:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-12-15 07:15:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-12-14 15:28:17.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-12-14 15:28:17.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-14 15:28:17.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-12-14 15:28:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-12-14 16:05:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-15 07:15:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-12-15 08:03:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abad64d3157bee17b583af2d10e1933a3238b7df","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-12-14 15:28:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f5dbd8590885f3e67103535056c78ebed2ab7dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-14 15:49:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/009fd2db6e004abdaa1c2601e4f1cd80 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/d6d499c741cd4916ae005d309b5d25c6 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/45392675be7a40c8a6fa09dfb904d714 : SUCCESS in 4m 36s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/af46c3401302469f9738d81357713489 : SUCCESS in 12m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8799b42c068e42659e49eefa840f5b7c : SUCCESS in 20m 24s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1670f191a3f54bcebc4f48cd77302a73 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68b2a7a93a30482d9d34c84222b2801d : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8dec1a1b171bcdc2072ec2f3f0fa10a6fe776ff7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-14 16:05:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bfd83b9626f8e0f729bafa01ad7458a825222333","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-12-15 07:15:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56c97b5890d7f041671b2cfadf56513b61c0ee9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 07:15:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d52f3e4c9847adde1b493fa4c8871714922eff06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:03:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"23e3984ee254a5d4e9bc7fd1beb46c08d5edcf32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:03:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/bd4f11a7090c41f3b15343d4ade28889 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/071dc8ffe57043048ec4d7793b22331b : SUCCESS in 5m 51s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/6010545c302b4c7b85cb813e924ebeab : SUCCESS in 4m 50s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/37eecfe4422f4ed0ab502b321851041d : SUCCESS in 18m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5551d0d608484d9992dcd58eac065c71 : SUCCESS in 28m 27s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ae41bd98e4f465f9a9e1e9e5994415a : SUCCESS in 17m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e2afeaff56945edbca3675c6321ad56 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"68c65cd576982beb43c407f3bdac283b91fb5a2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 08:04:47.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d69f0dca3e544189c38c521ae697e70 : FAILURE in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45524c504d7caac9ff09998e5322242660d8ed0a","revisions":{"45524c504d7caac9ff09998e5322242660d8ed0a":{"kind":"REWORK","_number":1,"created":"2019-12-14 15:28:17.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/69/699069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/69/699069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/69/699069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/69/699069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/69/699069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/69/699069/1"}}},"commit":{"parents":[{"commit":"27c80b158260be74ce7b5cf6d511d40b27a5f881","subject":"Add parameters for amphora anti-affinity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/27c80b158260be74ce7b5cf6d511d40b27a5f881"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:28:17.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-12-14 15:28:17.000000000","tz":0},"subject":"Revert \"Add parameters for amphora anti-affinity\"","message":"Revert \"Add parameters for amphora anti-affinity\"\n\nThis reverts commit 27c80b158260be74ce7b5cf6d511d40b27a5f881.\n\nChange-Id: I3c8dda3bbf8a09296fbe9ea3b7ae92b46dadf94d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/45524c504d7caac9ff09998e5322242660d8ed0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/45524c504d7caac9ff09998e5322242660d8ed0a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
