)]}'
{"id":"openstack%2Fpuppet-octavia~707434","triplet_id":"openstack%2Fpuppet-octavia~master~I8c2cfad24a6a526fdb667c57c6b75eb9c9f2da3d","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1861886","hashtags":[],"change_id":"I8c2cfad24a6a526fdb667c57c6b75eb9c9f2da3d","subject":"OVN Octavia Provider: Use SSL/TLS protocol when TLS is enabled","status":"ABANDONED","created":"2020-02-12 17:19:41.000000000","updated":"2020-02-25 19:06:08.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b9e9d7594e65d667a0de1c810028745626cb56a","_number":707434,"virtual_id_number":707434,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-14 13:41:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":-1,"date":"2020-02-14 14:22:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2020-02-14 13:12:10.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 15:47:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-02-13 17:09:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-02-13 17:14:49.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-02-14 13:41:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-14 14:22:07.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"}],"messages":[{"id":"e21ed550d8c7a247010a72f2d2c36b4802af1891","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-12 17:19:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d77afd49497be4a92527994d96ff442e8f3cacf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-12 17:21:34.000000000","message":"Topic set to bug/1861886","accounts_in_message":[],"_revision_number":1},{"id":"ed90f5d3d87212e02e6bdce2aab8702575e8a689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 17:48:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1823da90e7f484f87f2a6dcadb6a0c2 : SUCCESS in 7m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f0d77c827d14b7abe9ce5928eca7e2f : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/039d74a6a9ff401c87b387e220b493d6 : SUCCESS in 6m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f018a2bd1aa94b17a4979e8ba5ff771d : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8df396c287a44854ba0b4085f3ca98d7 : SUCCESS in 12m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3552c3f7e1d9403a8d5b743d408bc822 : SUCCESS in 12m 07s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/19d1e5b94b914b01a7d0d887a144e64f : SUCCESS in 26m 47s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3f4087df9e641a4a66b11aea3f80f4e : FAILURE in 15m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2372c1437d86491abefb292f3bd6f40b : POST_FAILURE in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bb7a21a916e7ef8b6185975ef0766085381565ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-02-13 17:09:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2cf0d3b8fea4a072b1e090851adac83f339a20cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 17:14:49.000000000","message":"Patch Set 1: Code-Review-1\n\nIn agreement with Michael\u0027s review. Would be nice to explore a different way to address provider driver-specific configurations. One option is to consider creating a separate .pp file (or even puppet project?) where the configurations would be generated to its own config file. The installer would then launch the API service with the provider\u0027s configuration file loaded, e.g. --config-dir /etc/octavia/conf.d/","accounts_in_message":[],"_revision_number":1},{"id":"a23e0ba8d81ebc7d26d81800f06615ee293898d1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 17:15:29.000000000","message":"Patch Set 1:\n\nFor reference: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master/deployment/octavia/octavia-api-container-puppet.yaml#L240","accounts_in_message":[],"_revision_number":1},{"id":"201faae632305a9316fdb28def9c56bea00f7d64","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-13 17:16:52.000000000","message":"Patch Set 1:\n\nOops! That line is for the driver agent but you get the idea :)","accounts_in_message":[],"_revision_number":1},{"id":"fadd275749c3f96a7418df0b82ea0961956417d0","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-13 18:44:45.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"eb96fa0cb942215fe02cce20eb1792dae3ad2bc1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-13 18:48:34.000000000","message":"Patch Set 1:\n\n\u003e Oops! That line is for the driver agent but you get the idea :)\n\nack. TY Carlos! Will work on this.","accounts_in_message":[],"_revision_number":1},{"id":"eab4ab72102853f07bfbca7a5142353dfb240481","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-13 18:56:17.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2923d593f65bbb2de6401258496ce71586392f5e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-13 23:09:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a49dcc354a4bee4741baee57d2b6a6aecf59bd77","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-14 11:20:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"35a3c9546acab6d5e0c51f2dc32cb37ad1534b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 11:48:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54dc4d11f7bd4c4fb59e0219e325ffe9 : SUCCESS in 6m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6077771bf1264eb28b91fb6a62b96ba4 : SUCCESS in 7m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0650c8c692e44b149520e7ed99d5f44f : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dbba610d77b64e668d63db77f05996d5 : SUCCESS in 20m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d4ad634bae74b1f83771c108ac7d72d : SUCCESS in 12m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/865e2547c46a40ef806115cad5d79cc3 : SUCCESS in 12m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1f43f0103ac5423abc81f98aa3e8380e : SUCCESS in 25m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23d959a293e741cf95259e6d7a301ef0 : FAILURE in 15m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d58e526f54248e385c0853f8c32a7ea : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2b0cff236733ef27f9b50f361f3fba1216331116","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-14 13:06:57.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d57c4f995905ab7f4a58e15ac845a634ab2a7b58","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-14 13:11:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c806001ab7b6d420790d32ec012df164a7101ab","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-14 13:12:10.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"a8e5386af7ca577d3ff1762639c16b0c7af44edb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 13:41:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66962d62fdd14486ab9b5537c2e7305a : SUCCESS in 5m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ec426e6a63448f399a86d7929ecd469 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1bf3d2de77a4ac68a881634667fe45e : SUCCESS in 5m 32s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/de52939a05814ad282bd74dbf7e80370 : SUCCESS in 19m 01s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a31a0550dbb34c0dbf87b0b5d710f094 : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04fc6d14018a403784534d480a71ed03 : SUCCESS in 13m 22s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7138223772ec48eca25073a9beb05c56 : SUCCESS in 24m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bab1980808de4f639f9e8723769d3019 : FAILURE in 15m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a39f6abfead142c695d5227e28b6d022 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b013e8dbed2965c218cc06b17163411b81d2e644","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-02-14 14:22:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b9e9d7594e65d667a0de1c810028745626cb56a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 19:06:08.000000000","message":"Abandoned\n\nAbandon. Will be using https://review.opendev.org/#/c/709618/ instead.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8124678922a22198b935da1b56abf5407ee102ee","revisions":{"f3cde5b064a27737e8e4e1873f05dac19229fe48":{"kind":"REWORK","_number":1,"created":"2020-02-12 17:19:41.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/34/707434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/707434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/1"}}},"commit":{"parents":[{"commit":"e5a5219dd98ba3989575b0c763b5f7460cf3bd45","subject":"Revert \"Add parameters for amphora anti-affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-12 17:17:50.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-12 17:17:50.000000000","tz":-360},"subject":"octavia: Use SSL protocol when TLS is enabled","message":"octavia: Use SSL protocol when TLS is enabled\n\nThis patch enhances Octavia config, so it can connect to\nOVN_Northbound DB using SSL.\n\nChange-Id: I8c2cfad24a6a526fdb667c57c6b75eb9c9f2da3d\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f3cde5b064a27737e8e4e1873f05dac19229fe48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f3cde5b064a27737e8e4e1873f05dac19229fe48"}]},"branch":"refs/heads/master"},"00aa920fad7641604507bced4285a030348f1101":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-14 11:20:02.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/34/707434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/707434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/2"}}},"commit":{"parents":[{"commit":"e5a5219dd98ba3989575b0c763b5f7460cf3bd45","subject":"Revert \"Add parameters for amphora anti-affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-12 17:17:50.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-14 03:18:19.000000000","tz":0},"subject":"OVN Octavia Driver: Use SSL/TLS protocol when TLS is enabled","message":"OVN Octavia Driver: Use SSL/TLS protocol when TLS is enabled\n\nThis patch enhances OVN Octavia config, so it can connect to\nOVN_Northbound DB using SSL/TLS.\n\nChange-Id: I8c2cfad24a6a526fdb667c57c6b75eb9c9f2da3d\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/00aa920fad7641604507bced4285a030348f1101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/00aa920fad7641604507bced4285a030348f1101"}]},"branch":"refs/heads/master"},"8124678922a22198b935da1b56abf5407ee102ee":{"kind":"REWORK","_number":3,"created":"2020-02-14 13:11:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/34/707434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/707434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/707434/3"}}},"commit":{"parents":[{"commit":"e5a5219dd98ba3989575b0c763b5f7460cf3bd45","subject":"Revert \"Add parameters for amphora anti-affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-12 17:17:50.000000000","tz":-360},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-14 05:22:40.000000000","tz":0},"subject":"OVN Octavia Provider: Use SSL/TLS protocol when TLS is enabled","message":"OVN Octavia Provider: Use SSL/TLS protocol when TLS is enabled\n\nThis patch enhances OVN Octavia config, so it can connect to\nOVN_Northbound DB using SSL/TLS.\n\nChange-Id: I8c2cfad24a6a526fdb667c57c6b75eb9c9f2da3d\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8124678922a22198b935da1b56abf5407ee102ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8124678922a22198b935da1b56abf5407ee102ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
