)]}'
{"id":"openstack%2Fpuppet-octavia~707936","triplet_id":"openstack%2Fpuppet-octavia~master~I4fc556fad4f5051a7a0272f1cb2001afa6d0b05e","project":"openstack/puppet-octavia","branch":"master","topic":"octavia-sa-region","hashtags":[],"change_id":"I4fc556fad4f5051a7a0272f1cb2001afa6d0b05e","subject":"Add region_name for service_auth","status":"MERGED","created":"2020-02-15 12:01:22.000000000","updated":"2020-02-16 23:55:41.000000000","submitted":"2020-02-16 23:54:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707936-1581897257232-824f8b14","meta_rev_id":"0251ef0a97a3f5d6a9e7664ca4f53ea2b8d64a28","_number":707936,"virtual_id_number":707936,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-16 23:54:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-16 23:24:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-02-16 20:34:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-16 23:24:12.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 20:34:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-02-16 23:24:12.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-16 23:54:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3dfb5dedeeedaa6517287944fa228a3f1e77e0a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-15 12:01:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23335486d4012d10b3ee781d9cf4e1036bbbb244","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-15 12:07:32.000000000","message":"Topic changed from octavia-service_auth to octavia-sa-region","accounts_in_message":[],"_revision_number":1},{"id":"82edb7b394c5f454d13323c7963c4e07033fb5cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 12:36:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae8ca14865f149a0855984af5a17a1df : SUCCESS in 6m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/115f8dd662cd4be38694e6e4b3db4345 : SUCCESS in 5m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a37ec3445af48b3966dbd7ef9bcee9a : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/423cea153cbf4ffbb56df13ef92b3155 : SUCCESS in 11m 17s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfe728817a9446e382205ca188597bc3 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a69ccf3f1ebf475998e78650cde4c015 : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7224c95299264679b02eaeb91c9253b5 : SUCCESS in 30m 39s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e13af8eab7604f5fa1422ca8a0997dcd : FAILURE in 18m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffae2e8a05aa48b8a91599953c4c4c4e : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"10db29970de47af0a5f272f137c52e6b9fb5dc61","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-16 20:34:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09bbfbe1281449d48232e6e533d03a439b0b825a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-16 23:24:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5548f255645f0e2a88c40bc0eae1fc3c3d88b775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 23:24:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6be9962c9da79d1a7fe68f850d5d994ec2233c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 23:54:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4e79ef1621c659e43218c9e96aedac24cc45f059","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 23:54:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2531351178942e7bcb9425a78ec57f7 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/763b8991782d4fbb9105326531923753 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/088603b46831490e8df0c1b8cb33ff8b : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f7cedf9e813b423d86798cc8eb7007a5 : SUCCESS in 19m 06s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27335febbe9f4ddc897d81d1e205d38a : SUCCESS in 12m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4cfb4753ef6d4f9b878df61cb972f3ce : SUCCESS in 25m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0eb8aecd60742e1a6fbdb21ba604d0b : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"0251ef0a97a3f5d6a9e7664ca4f53ea2b8d64a28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 23:55:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba93056034ad4142a603f857bae2716d : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43ab03de72fb0cffd3e9e13efdd8fbdc2fa615ee","revisions":{"43ab03de72fb0cffd3e9e13efdd8fbdc2fa615ee":{"kind":"REWORK","_number":1,"created":"2020-02-15 12:01:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/707936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/707936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/707936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/707936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/707936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/707936/1"}}},"commit":{"parents":[{"commit":"e5a5219dd98ba3989575b0c763b5f7460cf3bd45","subject":"Revert \"Add parameters for amphora anti-affinity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a5219dd98ba3989575b0c763b5f7460cf3bd45"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-15 11:58:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-15 12:01:05.000000000","tz":540},"subject":"Add region_name for service_auth","message":"Add region_name for service_auth\n\nAdd region_name for service_auth configuration, so that we can specify\nthe keystone region where octavia user is created.\n\nChange-Id: I4fc556fad4f5051a7a0272f1cb2001afa6d0b05e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/43ab03de72fb0cffd3e9e13efdd8fbdc2fa615ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/43ab03de72fb0cffd3e9e13efdd8fbdc2fa615ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
