)]}'
{"id":"openstack%2Fpuppet-octavia~709618","triplet_id":"openstack%2Fpuppet-octavia~master~I5922eba799eda860926acdb353a6ca275a375e5b","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1861886","hashtags":[],"change_id":"I5922eba799eda860926acdb353a6ca275a375e5b","subject":"Add Octavia OVN Provider configuration","status":"MERGED","created":"2020-02-24 21:47:19.000000000","updated":"2020-03-04 15:23:00.000000000","submitted":"2020-03-03 15:15:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709618-1583248535739-3bf77f9f","meta_rev_id":"ce7da39773147d834d3919b19b79ce70f19bb837","_number":709618,"virtual_id_number":709618,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 15:15:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-03-02 20:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":1,"date":"2020-03-03 11:22:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-03-03 11:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-03-03 14:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-03 10:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-03-03 14:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 15:47:48.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-02-25 21:10:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-03-02 18:30:43.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-03-02 20:09:09.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-03 10:44:02.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2020-03-03 11:22:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-03-03 11:36:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-03 14:45:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-03 15:15:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bc2f3a34bd0574fa4879b316bbc165b67f13f85","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-24 21:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abc5f40083d17e9333414e77292e6d9dc1076303","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-24 21:47:55.000000000","message":"Topic set to bug/1861886","accounts_in_message":[],"_revision_number":1},{"id":"c7505dcf899e72ab4da3bee6d9dd978e0ae4a03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 22:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44a3a5208b1e40bcb2329b1dee37fffc : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1096b51b341545e8844df819415a3ffe : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1dc936919074b109f4d92c578be5be1 : SUCCESS in 5m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/107e4c96c1394d97874e5ff7afbef1ba : SUCCESS in 15m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36de8a20649f4b47af363767e72315a0 : SUCCESS in 14m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/474be71a5f89433bbc9867164c70c428 : SUCCESS in 15m 57s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/24d31297ddd743c9b6ece37fde6faea6 : SUCCESS in 31m 32s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa65e52d99e240e3a7a5d939eabf2f53 : FAILURE in 18m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e368aea4dd84bf09af7d930edabb3b6 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e3d728941460f67be37d8decef4c8575f7ea55a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-25 07:14:56.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"803cc3905b41c2b30a7aadd248fb793989970030","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 10:05:08.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nTY for the comments, Takashi!","accounts_in_message":[],"_revision_number":1},{"id":"39759fe8ae1281cb92d3459366f1c85984e09605","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 10:05:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4723745147615113173744217d8264162b99a9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:41:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4975678d45f42c8b6bbfb7eacc36f8a : SUCCESS in 8m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/788d927479da48a68ae4233906824082 : SUCCESS in 7m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b694dfcd2afd48f083db4f08c6a95d23 : SUCCESS in 6m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7d55964a699f4f119635d0a99ba10c4f : SUCCESS in 16m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7603708902c4ffdb07df29982303915 : FAILURE in 18m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2c9befb4e004ac2be0d490cd200c230 : FAILURE in 18m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bbec727f483a4b8caedd3683ae54967d : SUCCESS in 31m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05f857f754f3488dac87f55c0ae95cc4 : FAILURE in 17m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5988e2e66f6e465fb834bd2de2906e9a : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"98ebef60d7e395811710439a55d17e36891b3fb6","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 10:47:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"762334283cde6a17f57c5abd0e54ee34e92f6f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:24:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a35eacf3127740978f8aff541e9d7d7b : SUCCESS in 9m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bbcfa3bf08c460187850897476a416b : SUCCESS in 7m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01ee5e38783748f69228cf6c07520d53 : SUCCESS in 7m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6e5d1c3a69b64f288fce8bb78931dbe1 : SUCCESS in 21m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58accd0bb0d345f0afadaf27fc4a4dc7 : SUCCESS in 16m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/616d570e04d940d98fbab743c2fc1013 : SUCCESS in 15m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7839b7f354d14a019884ee096ea65ee1 : SUCCESS in 31m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93342cc86c5841359e389664f5b2d855 : FAILURE in 14m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd9142d2ad6a4377921c32e7de9585bf : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f701ae432a7a4777aec1f45834c97d75e4187f91","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-25 11:45:20.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM. Thanks!","accounts_in_message":[],"_revision_number":2},{"id":"095c7c5efdbde5615895724ab2237a8a689d68dc","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-02-25 15:36:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65192d54fde00232b6fee314b8d13aa9e9cb1278","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-02-25 15:47:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d17d0bb68486b09ad96bfeacdb007306678c1d88","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 15:50:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3047728668c13d3d826360d03845caf72babf8e2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 15:51:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e916b8e7e675f550a6fb59cbd831dade31e54f3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-25 15:51:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f90fcc98bf65c7995b8784699759d97ae31180e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:20:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3da383dcdabb4a66a56b26385ce27c6b : SUCCESS in 5m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff7586d0185f41519c7f438fa15b9d0f : SUCCESS in 6m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d29d48c898e464686144b568f8199bb : SUCCESS in 6m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d9a08dc4082c49ada6e28fd412b5d32e : SUCCESS in 17m 01s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d66b6fd53ab475791448e63ea623c9a : SUCCESS in 13m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1dcb8c9dd0d47928ca85f4ca6a835d1 : SUCCESS in 12m 37s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d157870bf22b4d3c8e4018c8e0418466 : SUCCESS in 27m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c7b9aa7d1b4de3960ad517541ea556 : FAILURE in 18m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c656c4a43b87424f800e3b89d9fabe0a : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e9505ac55fec449eff65563f8ff690fc3154fb9a","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-02-25 17:56:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"71639e6a733f35a510fde7c1b2839730dfd4ae53","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-25 21:07:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fa4825c74bc03e4b6818d72117940dfbd03abd6e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-02-25 21:27:07.000000000","message":"Patch Set 3: Code-Review-1\n\nDup of https://review.opendev.org/#/c/663393/ for driver agent configuration.\n\nAlso, the Octavia driver agent does not have such configuration options. This is OVN provider driver-specific.\n\nIn another change, there was a recommendation not to pollute driver specific configurations with Octavia services. A better design is to place provider driver-specific configurations in a separate file.","accounts_in_message":[],"_revision_number":3},{"id":"47d40870e5ef4c06fe40cadf14c20f49f4e4b8d3","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-02-26 15:12:02.000000000","message":"Patch Set 3: Code-Review-1\n\n\u003e Dup of https://review.opendev.org/#/c/663393/ for driver agent\n \u003e configuration.\n \u003e \n \u003e Also, the Octavia driver agent does not have such configuration\n \u003e options. This is OVN provider driver-specific.\n \u003e \n \u003e In another change, there was a recommendation not to pollute driver\n \u003e specific configurations with Octavia services. A better design is\n \u003e to place provider driver-specific configurations in a separate\n \u003e file.\n\nWhile I was thinking of doing this a bit more incrementally, I think we are just going to avoid the confusion and go right for the jugular.\n\nSo I think what we need to do here is to do the following:\n - deprecate the misplaced ovn_nb_connection that is in api.pp in a backwards compatible way (lest we break CI for the current provider job)\n - add a config file provider for ovn.conf to create the new config file, the location would be some subdirectory under the service (see neutron ml2 configurations for examples)\n - create a directory named \"providers\", and add an ovn.pp for the ovn related params and have it use the config provider\n\nIn tripleo heat templates:\n- alter the command line for the driver agent to pick up the new file. Carlos\u0027 suggestion was to use config-dir to allow the driver agent to pickup multiple drivers (can the driver agent actually do that?)","accounts_in_message":[],"_revision_number":3},{"id":"0510b18c9f2088bde71656863cd96728321091bc","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 04:32:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0061af49c6f693aa49eada36c2961a6535288560","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 04:34:13.000000000","message":"Patch Set 4:\n\n\u003e Dup of https://review.opendev.org/#/c/663393/ for driver agent\n \u003e configuration.\n \u003e \n \u003e Also, the Octavia driver agent does not have such configuration\n \u003e options. This is OVN provider driver-specific.\n \u003e \n \u003e In another change, there was a recommendation not to pollute driver\n \u003e specific configurations with Octavia services. A better design is\n \u003e to place provider driver-specific configurations in a separate\n \u003e file.\n\ndone.","accounts_in_message":[],"_revision_number":4},{"id":"e027ac1960bce9c9ea99edec9f4ff59e7f5324cf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 04:34:34.000000000","message":"Patch Set 4:\n\n\u003e \u003e Dup of https://review.opendev.org/#/c/663393/ for driver agent\n \u003e \u003e configuration.\n \u003e \u003e\n \u003e \u003e Also, the Octavia driver agent does not have such configuration\n \u003e \u003e options. This is OVN provider driver-specific.\n \u003e \u003e\n \u003e \u003e In another change, there was a recommendation not to pollute\n \u003e driver\n \u003e \u003e specific configurations with Octavia services. A better design is\n \u003e \u003e to place provider driver-specific configurations in a separate\n \u003e \u003e file.\n \u003e \n \u003e While I was thinking of doing this a bit more incrementally, I\n \u003e think we are just going to avoid the confusion and go right for the\n \u003e jugular.\n \u003e \n \u003e So I think what we need to do here is to do the following:\n \u003e - deprecate the misplaced ovn_nb_connection that is in api.pp in a\n \u003e backwards compatible way (lest we break CI for the current provider\n \u003e job)\n \u003e - add a config file provider for ovn.conf to create the new config\n \u003e file, the location would be some subdirectory under the service\n \u003e (see neutron ml2 configurations for examples)\n \u003e - create a directory named \"providers\", and add an ovn.pp for the\n \u003e ovn related params and have it use the config provider\n \u003e \n \u003e In tripleo heat templates:\n \u003e - alter the command line for the driver agent to pick up the new\n \u003e file. Carlos\u0027 suggestion was to use config-dir to allow the driver\n \u003e agent to pickup multiple drivers (can the driver agent actually do\n \u003e that?)\n\ndone.","accounts_in_message":[],"_revision_number":4},{"id":"8ce2c7dc4bbb7abde9b9e72a94e11e92cabdbaac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 05:01:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4748fc183297432a8f353c6f23c58b00 : FAILURE in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a09a7adf55647e5910799bf7551312a : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b2312f8231242c6b8b0c79d51b0a6a7 : SUCCESS in 5m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71337d34038e4b7e85ec4acb8430de6b : FAILURE in 20m 10s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8afcb3b3b2744a968710812843b1cb57 : FAILURE in 11m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3b8629f3dca4abe8ce7c5f618e26b5f : FAILURE in 10m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e17e0d37e0e141d396edb61de5074ad3 : SUCCESS in 25m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8d7108c1ade46718a3062cdaa5c21b0 : FAILURE in 16m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c524ed6305954052a3b95d42ae3fa126 : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"33a3064be08adf3ed17f050689efbec78e53e96c","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-02-27 13:19:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ba3b61d687df2efbfeefc85ffa62d427535a77a3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 15:06:36.000000000","message":"Patch Set 4: Workflow-1\n\n\u003e (1 comment)\n\noops.. definitely :facepalm","accounts_in_message":[],"_revision_number":4},{"id":"ed45e2020b17b0e2261ccdeb6eb1131b92f4d5d2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 15:07:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"be750f7e6315013d3ea12bfc6ee870ce47c176c7","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2020-02-27 15:42:11.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8d7fd09cc97a10d40285b42b1a9f4bf973d16ba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:46:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9555dedc19894eceaefdf56ad109ba2f : FAILURE in 7m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e4cf0fd7d324cdfb9081f9527a2c42d : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/edaed337e3844224b2bf9353a1c1e292 : SUCCESS in 7m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/64bd405583a8445689d321592a5150a3 : FAILURE in 17m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c519c9408a76451887d4f4f9ca757d90 : FAILURE in 13m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75d33d8c5a934f34bc48e29181a4a59f : FAILURE in 13m 18s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e99f1def1f9c4eccb297f3d90afbca85 : SUCCESS in 27m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/361097c4dc774ef189c37249a8607b92 : FAILURE in 16m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66281b0306354c358835e0a1ae53c9a3 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d8367ff2086ecbb5fd385092f3d73bd0e438b87d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 16:00:03.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"944e044df665687d32f81d5b64898e19ea87e580","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 16:00:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b9d27983618e4c259b6ffbe541a3fe194cc77142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:12:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c62a43553352493692fad45354b09cc2 : FAILURE in 8m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c58192e99ea4eb183d68e3621714e0f : SUCCESS in 7m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f878426d0aa547c99233e5570b505f7f : SUCCESS in 6m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/605ad65dd1064d1fb25cabd572f86150 : FAILURE in 21m 43s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69c105a117ac4b8682f6f87fbe332d10 : FAILURE in 14m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46bc291aa1ba492ea7a327f895bca9dd : FAILURE in 16m 57s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7da251274db24d08894aecff25e0cddb : RETRY_LIMIT in 26m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ac175cac02a451884e5a243601b5f83 : FAILURE in 19m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41d2d4b1ab8048f090eec2353aa8a944 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":6},{"id":"045d39faf93d00507b6339abbcd826c7d9166831","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 19:23:57.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"c75217e789c7159c43a8fa9144b2f9c231ac19a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 19:54:22.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46ef7d5211c54d2480ac3ee33f9ad3a9 : FAILURE in 5m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7dd868e5f460469e90d9263f3c9b2ce8 : SUCCESS in 7m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56e9d882e2794aba8182c3bc7bee260f : SUCCESS in 6m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/194f01fa961a4444b0367ab3a02cae4b : FAILURE in 12m 46s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d9890c5a49b426e87ca1552bebb09d5 : FAILURE in 11m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dcae43c09844b4d81ebc5c320239f9d : FAILURE in 12m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3bff760cf9104e1fad34ecfabf94b01a : SUCCESS in 28m 08s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53407d5595f844d4b7f24a9de35c85a3 : FAILURE in 15m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/121860e837e54b729113e51dcf37394f : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":6},{"id":"0d4c2fe68007c6828a517a4987dd434f8bce37a3","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-27 20:30:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7f323116e522e2f97e336802ff79b3e6da90f2d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 21:15:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a2cb6f2fd334a759a94667a3ebec50a : FAILURE in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2ba2a5b941b4f88b7191caa13fb541d : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c5b9e70570c475b9fe3204b7f87648d : SUCCESS in 6m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bd36a6151e144c9aa9dfa5e3cbaf6c0f : FAILURE in 12m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56175921a0184f8db2e1ad68f4a8a6c6 : FAILURE in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bae9b4de89a4e8aa4a54ccdfe841e2f : FAILURE in 12m 12s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7e7c5774e97f499996499e78f1f4a886 : SUCCESS in 32m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/465d82268d9c426083a5793d25b8c7fc : FAILURE in 14m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955e15f660d94248874c857429f9600e : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":7},{"id":"cd1926b4d24cd9e61eeaaef09b5ea5b336035e78","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 10:48:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"92f6a9180944ffa67940419748be9a724b2b523e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:19:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4729a7380707448790df24ed0b8611fb : SUCCESS in 6m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/527fd1d02b6343e6afcaa3a4b1f2e66c : SUCCESS in 6m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbfbaeae8386422db55a47993141dab9 : SUCCESS in 6m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e433674a94f443d293ec2d1e70aed307 : FAILURE in 20m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86bff9d6e14e4048a199b87d04af1df1 : FAILURE in 16m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4a56961874484c8b83c605042f17a859 : FAILURE in 12m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/13e60cc0d6964b3c9f000a27b6c86f0d : SUCCESS in 25m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/914068d26ee44688a947df8b1bc8f89e : FAILURE in 14m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a786b49ac77472ea5190fb185f1c670 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":8},{"id":"97f7486c719b779381cc52b97336f4335e3b2f50","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 11:25:55.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2fa1b23cf6ab08ef0d71cf48282eca159a2df8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:59:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15a129cb3af9478887c3c903c90696d2 : SUCCESS in 6m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a25a60cb0ce14591bea263116c8cf05c : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bcee8d294974bb0b25343e588e950d4 : SUCCESS in 6m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d148bc2ccb3946b2a9872b6ae22224b6 : FAILURE in 21m 37s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d64b9a86da994756951f950c28f3428d : FAILURE in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b4b9f294ab343caba55f3779c72cb4e : FAILURE in 11m 27s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ca3e771a16814a40823b677cfe06c500 : SUCCESS in 26m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fc7e54c13b945408f7e53a513c4b962 : FAILURE in 15m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e59e54335a41c3a00335544c72a5ed : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":9},{"id":"98ccde2ced9a871c334f7e89e2d38d51cb7e72f1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 12:16:19.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ded80d09661538dd3411de389d0579684a07f015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:44:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1af84356d4eb4b0e8c3cdcb81aa4c6df : SUCCESS in 5m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a695c19c4a9945389867c0d079275fff : SUCCESS in 5m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf82eda37c5f48328e4e7b9a5255712a : SUCCESS in 5m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c8128dbbc1934322b36843a073adaf87 : SUCCESS in 13m 40s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fd5fe5d695c4bbdb379aaa815ff3a22 : SUCCESS in 12m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f90317ed36564bb59585c31a907a0e20 : SUCCESS in 12m 28s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2a55e5a7985445588fe651e812dea635 : SUCCESS in 25m 04s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09a76e7a595d4f2ea9920d58001eaf87 : FAILURE in 15m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5cfd1523404028bf26bbc98b12e066 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":10},{"id":"9b973c855e4e846ece781f8b7474c43f5436686c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-28 13:22:07.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)\n\nmissing testing for new provider/type, release note doesn\u0027t include the deprecation and two small linting","accounts_in_message":[],"_revision_number":10},{"id":"f74cad5d938623010a2c15d372f1de574d94f9a7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 22:24:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0f0ed240ac18279de6e368890c954994fba88118","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 22:27:26.000000000","message":"Patch Set 11: Workflow-1\n\n(2 comments)\n\nThank you, Tobias.\n\nI\u0027m doing something wrong in spec/unit/type/octavia_ovn_provider_config_spec.rb and need to revisit that.","accounts_in_message":[],"_revision_number":11},{"id":"5cc6786083e6b192ab61e919c3cd1581187e42ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 22:59:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e9aaa7640bf4ea095bf325f346f5cae : SUCCESS in 6m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e0dce6766f14a30a3048ccc0096821e : SUCCESS in 5m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c34a414be1fb42ae9ab92b81f6e6b073 : SUCCESS in 6m 45s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e6809ac068cd4b84bdd0c911889b5ed4 : SUCCESS in 20m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1357eb70f182479db5c4e95a34a7689f : SUCCESS in 13m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6373801b0a224d308fdfef9a27024c41 : SUCCESS in 11m 03s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/504eb80c4b824de9b0497da6657799ba : SUCCESS in 32m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/939e42deaf7c478fabde2199a56bc70b : FAILURE in 14m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79e753fb8e284825a72833cccba2587a : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":11},{"id":"ce1a9ae5085aab1da86851b1d3ac94a58e5e4b5f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-02-28 23:24:03.000000000","message":"Patch Set 11: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2bcf1521a94bc9cc1ab4319a299b7008a4a7b914","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-01 13:44:10.000000000","message":"Patch Set 10:\n\n(5 comments)\n\nAlmost looks good to me, but it would be nice if the comments about default value and deprecation warning are addressed.","accounts_in_message":[],"_revision_number":10},{"id":"6120694ad7108d8e04717c07b2ca02c8ffaa678e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-02 11:31:20.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a8eb614bb4da6461824fce0ecd1b5f34645a4751","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-02 11:32:04.000000000","message":"Patch Set 11:\n\n(5 comments)\n\nThank you for the review, Takashi!","accounts_in_message":[],"_revision_number":11},{"id":"44a9822004ecd08658c0835e17cb3c89c1122b2b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-02 11:45:35.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"7c6fdb32412bc388b819339f15e208cb95e5048f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:59:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13555939282647768165b957c907ffa3 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cc9ca1914254db6ae18e9418418279b : SUCCESS in 6m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04c3c49c82ef4c409941d5e6d32c8d7b : SUCCESS in 6m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/51132a98f6a14e2db4cef44e658a1060 : SUCCESS in 21m 47s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dad1196a16f4438b8b3de26c6b2be6f : SUCCESS in 11m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/086c4cd1a359451a8b70c377d810e783 : SUCCESS in 13m 10s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ba402ba22e274b398b20c6572401d6cd : SUCCESS in 25m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3344d2bb29744081b7bc187254c3a9db : FAILURE in 18m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/556b5c9a4cae4f689487b67cd3d30855 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":12},{"id":"f762aca8fb3a546299a930e205a07d1582ad45ae","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-02 14:51:46.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"e553535f4630f8eaf039918fc2395eb0ec716ee9","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-02 18:30:43.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"83d20aa17bbd0ae65bc5219cfc69c1c3f6b84d5b","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-02 18:38:20.000000000","message":"Patch Set 12:\n\ncgoncalves should probably +1 before this merges","accounts_in_message":[],"_revision_number":12},{"id":"2b092eebb8a0036348f2214a58c10cc5f118b124","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-02 18:40:42.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)\n\nNits, otherwise LGTM.","accounts_in_message":[],"_revision_number":12},{"id":"596d8203d92dd781ab7bff355a2809093ebbe917","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-02 19:50:50.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e2090d2b768869874a194cd09a419bdfde063ae8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-02 19:51:04.000000000","message":"Patch Set 12:\n\n(2 comments)\n\nTY Carlos!","accounts_in_message":[],"_revision_number":12},{"id":"bde49be59f11966595cbc529701d96aaef7e0a9f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-02 20:09:09.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"479b307a833ad94439b5427ceb27c0637b7cdf00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 20:30:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5636d915066445db7a74e03b3d35877 : SUCCESS in 8m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b04bb35bcf84461a8dd7601c548f2446 : SUCCESS in 8m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8db82aa7814d4da5a95184d4979de9db : SUCCESS in 9m 03s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c4ebc5615bd44ec39f23f3730a9d542c : SUCCESS in 19m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c592d9c425b949a58828fe2241307c19 : SUCCESS in 14m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4db914e66a354fe8a9f35c2815cb7b8d : SUCCESS in 14m 27s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/30138b17e1424ea28e1c5aee8e5248d0 : SUCCESS in 32m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/182d08f8b5274be9a3b7db05218541e3 : FAILURE in 15m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2257605002e4382bd78431b1ec07e40 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":13},{"id":"57c1f011e5c3e16e39f755b494c04dd68ba8548f","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2020-03-03 10:44:02.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"45474d2712b261666e13128c0503e7fed439339a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-03 11:22:39.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"52beea6208f5b15285036803cb7124e2c9b2a6e5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-03 11:36:41.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"db7e21ee831d0eade314858ca4a1aa13268436f1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-03 14:45:32.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"8bc984a1e101304b6ff16137d7195368483ad0c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:45:47.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"3ee4d18a2653887b9ef6aaa8c694bc969fd8ab99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:15:35.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea44599dd1084eb584218f313688fa57 : SUCCESS in 6m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/403b92d71c524ae7bb06b246522ad602 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/775d38b5f70e4e058cbbab4bb3d81646 : SUCCESS in 5m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/590bc6eb7801490c8f80c4b2ff488950 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccacffc2f9e34194ae7323e01eec1bb0 : SUCCESS in 12m 46s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/537542350dec4cc9b4549c430af018da : SUCCESS in 27m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd688f521cd447d5be9c88398a13c28b : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":13},{"id":"ef4cc3068ea2d45130fd817cca267c7810408138","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:15:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":13},{"id":"3b3aaf2754ae42afa47801be18b3de10fe3d9098","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:17:16.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b630a869bef746c6b0a7ca3965f71288 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":13},{"id":"ce7da39773147d834d3919b19b79ce70f19bb837","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-03-04 15:23:00.000000000","message":"Created a revert of this change as I83489bfa2940018dd09639dcbc33ab6deabd593a","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"81b45f9eab723a47265bb9d73725b2793a052167","revisions":{"43e2c72329a4964ee4b38da94dffcce02f21fde4":{"kind":"REWORK","_number":1,"created":"2020-02-24 21:47:19.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/1"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:56.000000000","tz":-300},"subject":"Add driver agent configuration, starting with ovn TLS parameters","message":"Add driver agent configuration, starting with ovn TLS parameters\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nNote that there is currently an OVN attribute under Octavia API that should\nbe moved from class octavia::api to newly created octavia::driver_agent:\novn_nb_connection. That move will take place in a future gerrit change.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/43e2c72329a4964ee4b38da94dffcce02f21fde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/43e2c72329a4964ee4b38da94dffcce02f21fde4"}]},"branch":"refs/heads/master"},"146a5605de6f4b09a740d4cb41f1187bdf9bcdb3":{"kind":"REWORK","_number":2,"created":"2020-02-25 10:05:33.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/2"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-25 10:05:30.000000000","tz":-300},"subject":"Add driver agent configuration, starting with ovn TLS parameters","message":"Add driver agent configuration, starting with ovn TLS parameters\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nNote that there is currently an OVN attribute under Octavia API that should\nbe moved from class octavia::api to newly created octavia::driver_agent:\novn_nb_connection. That move will take place in a future gerrit change.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/146a5605de6f4b09a740d4cb41f1187bdf9bcdb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/146a5605de6f4b09a740d4cb41f1187bdf9bcdb3"}]},"branch":"refs/heads/master"},"0b3f6faa9aff5bfe97ac0b7cc13ae6e869fd38f7":{"kind":"REWORK","_number":3,"created":"2020-02-25 15:50:42.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/3"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-25 15:50:38.000000000","tz":-300},"subject":"Add driver agent configuration, starting with ovn TLS parameters","message":"Add driver agent configuration, starting with ovn TLS parameters\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nNote that there is currently an OVN attribute under Octavia API that should\nbe moved from class octavia::api to newly created octavia::driver_agent:\novn_nb_connection. That move will take place in a future gerrit change.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0b3f6faa9aff5bfe97ac0b7cc13ae6e869fd38f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0b3f6faa9aff5bfe97ac0b7cc13ae6e869fd38f7"}]},"branch":"refs/heads/master"},"3f1f1d4688d506063e2a4325e0101b247b1b12a8":{"kind":"REWORK","_number":4,"created":"2020-02-27 04:32:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/4"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-27 04:31:23.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f1f1d4688d506063e2a4325e0101b247b1b12a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f1f1d4688d506063e2a4325e0101b247b1b12a8"}]},"branch":"refs/heads/master"},"11c391ff0f089ae06feb1c51b70303f7601b7745":{"kind":"REWORK","_number":5,"created":"2020-02-27 15:07:47.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/5"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-27 15:07:44.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nCloses-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/11c391ff0f089ae06feb1c51b70303f7601b7745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/11c391ff0f089ae06feb1c51b70303f7601b7745"}]},"branch":"refs/heads/master"},"5f4a3cdc4738fdb783fba5286d7a4f2d1a997916":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-02-27 16:00:03.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/6"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-27 15:59:45.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f4a3cdc4738fdb783fba5286d7a4f2d1a997916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f4a3cdc4738fdb783fba5286d7a4f2d1a997916"}]},"branch":"refs/heads/master"},"95f38b8031614e56f26124e77f488b8d92e2d132":{"kind":"REWORK","_number":7,"created":"2020-02-27 20:30:47.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/7"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-27 20:30:42.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/95f38b8031614e56f26124e77f488b8d92e2d132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/95f38b8031614e56f26124e77f488b8d92e2d132"}]},"branch":"refs/heads/master"},"edafc8dd54bed9f950855b478327d9d1ac69377f":{"kind":"REWORK","_number":8,"created":"2020-02-28 10:48:10.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/8"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-28 10:48:06.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/edafc8dd54bed9f950855b478327d9d1ac69377f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/edafc8dd54bed9f950855b478327d9d1ac69377f"}]},"branch":"refs/heads/master"},"b0e87cfa625ef75c88831d4e8ae309ec418d88f5":{"kind":"REWORK","_number":9,"created":"2020-02-28 11:25:55.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/9"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-28 11:25:52.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0e87cfa625ef75c88831d4e8ae309ec418d88f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b0e87cfa625ef75c88831d4e8ae309ec418d88f5"}]},"branch":"refs/heads/master"},"871f1db02d46f3b813feae5482d5581d40533650":{"kind":"REWORK","_number":10,"created":"2020-02-28 12:16:19.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/10"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-28 12:16:14.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/871f1db02d46f3b813feae5482d5581d40533650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/871f1db02d46f3b813feae5482d5581d40533650"}]},"branch":"refs/heads/master"},"2a506fd76f12656e43675ae84281bfe6056e2faf":{"kind":"REWORK","_number":11,"created":"2020-02-28 22:24:45.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/11"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-28 20:25:27.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2a506fd76f12656e43675ae84281bfe6056e2faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2a506fd76f12656e43675ae84281bfe6056e2faf"}]},"branch":"refs/heads/master"},"593d42e4d5d6220c48054506de59c6b00c216387":{"kind":"REWORK","_number":12,"created":"2020-03-02 11:31:20.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/12"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-02 11:31:15.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/593d42e4d5d6220c48054506de59c6b00c216387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/593d42e4d5d6220c48054506de59c6b00c216387"}]},"branch":"refs/heads/master"},"81b45f9eab723a47265bb9d73725b2793a052167":{"kind":"REWORK","_number":13,"created":"2020-03-02 19:50:50.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/18/709618/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/18/709618/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/18/709618/13"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-02 19:50:44.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nChange-Id: I5922eba799eda860926acdb353a6ca275a375e5b\nRelated-Bug: #1861886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81b45f9eab723a47265bb9d73725b2793a052167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81b45f9eab723a47265bb9d73725b2793a052167"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
