)]}'
{"id":"openstack%2Fpuppet-octavia~711122","triplet_id":"openstack%2Fpuppet-octavia~master~Id26f6c1c252121f543938305098d22f6510299e5","project":"openstack/puppet-octavia","branch":"master","hashtags":[],"change_id":"Id26f6c1c252121f543938305098d22f6510299e5","subject":"Change the default for octavia-worker workers to os_workers","status":"MERGED","created":"2020-03-03 22:56:36.000000000","updated":"2020-03-09 15:31:03.000000000","submitted":"2020-03-09 15:29:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711122-1583767751737-72e87165","meta_rev_id":"84e80be14c58ae39e00d808b727e01aac990a3d0","_number":711122,"virtual_id_number":711122,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 15:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 14:56:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-05 10:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-03-05 00:45:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-03-05 09:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-09 14:56:56.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 00:45:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-03-05 09:06:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-05 10:36:50.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-09 14:56:56.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-09 15:29:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cea7789d32fbbf01e76daab9fae9facc56a2c1bb","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-03-03 22:56:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8d350bfa8e5df54a5176436b3dbc34bca65330a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 23:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f9c1700c41c4eafb702812222dfe434 : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da44527608aa4fbd8350f6c22be58393 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0071bdd33ecc460b8fee0d98b442e9a6 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/769038b9e901432f87c6e7a47881843a : SUCCESS in 13m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cff10031c39417ca60ba9f4b635062a : SUCCESS in 10m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7159f56e029f4a0fa402c392d58c1cef : SUCCESS in 10m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/28f79358d566497d9b561fbbe209652d : SUCCESS in 26m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01b2ab04524744389d952089256146bd : FAILURE in 15m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf48779df244db6b6686ab7eef11341 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1d4b68ae598835e19ad290091139d9cfd34f2b42","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-03 23:48:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b480cabcc6acf348a8ba85ec9a8dcc58b27d3a82","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2020-03-04 21:45:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84bbec0c8f952ae5c9de5ca3ae6a487e19c570d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 22:21:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2ad8e95470f4210b56a401fc9b9628e : SUCCESS in 6m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3eaf7c8441dd4618b2c4be4251942fa8 : SUCCESS in 5m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5a75d4b869d4c87ac111d4a44ef09c4 : SUCCESS in 5m 43s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c0aae1dc381847ce81acd204fc2125ed : SUCCESS in 14m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/764fd9887d1140399c0f6588fe0474f3 : SUCCESS in 11m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b3d220981ee4d689a63b7630758c695 : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4b5adaea8fab49e5ac02dca1763792cb : SUCCESS in 30m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5efbf9be662b42169048ccf6b1777150 : FAILURE in 15m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb482fb8b63b429eb4ce9881982cdc5d : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"684380402d08b579d57aed096f8f6cfc35c93634","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 00:45:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3999d7890514e028321ad6d6d8b56d53f569d3b6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-05 09:06:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c905f67b598711754ccc9e4906d7b3e4f50e87a8","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-05 10:36:50.000000000","message":"Patch Set 2: Code-Review+1\n\nJust a comment to note that the housekeeping class does not have a workers parameter. It could be added in this change or a follow-up one.","accounts_in_message":[],"_revision_number":2},{"id":"56acaaf5ea47066f82d73408e15676825c58e00d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 14:56:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0a087a4521ce2cd87ee32cfdf8914c648dfa106d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 14:57:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21b675b19f4815a943688e1f509d7aea50ec8656","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:29:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dfc5cf1d2854deb9899cfcca22a3e7c : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70599cce7f014f33b307727c4f719081 : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc0f7abd72fe42d9a4b46f8390b172ce : SUCCESS in 4m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8ad34caf855b4af9883e0f7d8519ca55 : SUCCESS in 17m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86f0671b88c449a595d0cefdbacc44a5 : SUCCESS in 11m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b730931081f94f20a103637072df74f0 : SUCCESS in 25m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02d4cb79822f40a385b8efa1809cc814 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e30945fce67009caac11e0e98e29c69373f39954","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:29:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"84e80be14c58ae39e00d808b727e01aac990a3d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 15:31:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a6b3879201e48d8b898b55e127083e6 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d8a4a86ed0c9b7d9033c8f2494f41fbacfe586f","revisions":{"49bcfe077a5e7c2781b740c3172add9fad910536":{"kind":"REWORK","_number":1,"created":"2020-03-03 22:56:36.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/22/711122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/711122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/1"}}},"commit":{"parents":[{"commit":"81b45f9eab723a47265bb9d73725b2793a052167","subject":"Add Octavia OVN Provider configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81b45f9eab723a47265bb9d73725b2793a052167"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-03-03 22:55:30.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-03-03 22:56:33.000000000","tz":660},"subject":"Change the default for octavia-worker workers to os_workers","message":"Change the default for octavia-worker workers to os_workers\n\nPreviously this was set to os_service_default but should use the\nos_workers fact\n\nChange-Id: Id26f6c1c252121f543938305098d22f6510299e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/49bcfe077a5e7c2781b740c3172add9fad910536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/49bcfe077a5e7c2781b740c3172add9fad910536"}]},"branch":"refs/heads/master"},"5d8a4a86ed0c9b7d9033c8f2494f41fbacfe586f":{"kind":"REWORK","_number":2,"created":"2020-03-04 21:45:33.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/22/711122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/711122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711122/2"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-03-03 22:55:30.000000000","tz":660},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2020-03-04 21:45:22.000000000","tz":660},"subject":"Change the default for octavia-worker workers to os_workers","message":"Change the default for octavia-worker workers to os_workers\n\nPreviously this was set to os_service_default but should use the\nos_workers fact\n\nChange-Id: Id26f6c1c252121f543938305098d22f6510299e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5d8a4a86ed0c9b7d9033c8f2494f41fbacfe586f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5d8a4a86ed0c9b7d9033c8f2494f41fbacfe586f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
