)]}'
{"id":"openstack%2Fpuppet-octavia~711322","triplet_id":"openstack%2Fpuppet-octavia~master~Id1bd4753a876cd7c41461fb28789f7c6f8679b3f","project":"openstack/puppet-octavia","branch":"master","topic":"centos-8-jobs","hashtags":[],"change_id":"Id1bd4753a876cd7c41461fb28789f7c6f8679b3f","subject":"add tripleo job for octavia, centos-8","status":"ABANDONED","created":"2020-03-04 19:47:27.000000000","updated":"2020-03-25 02:01:41.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70fe2100f8580de6333f9c14dbcc40124ad44a5a","_number":711322,"virtual_id_number":711322,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-10 13:00:41.000000000","_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-09 18:17:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":-1,"date":"2020-03-10 13:11:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 20:51:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-09 18:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-10 13:00:41.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2020-03-10 13:11:11.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"d7b93f86a8afd4c405d64a3457bde46db1094fc1","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-04 19:47:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1df3ab8004778d0dc26d5824849a2cce3919ed6b","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-04 19:48:20.000000000","message":"Topic set to centos-8-jobs","accounts_in_message":[],"_revision_number":1},{"id":"8e9b9432d3408ac2dbcdbb7c4d853f1f4e15013a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 20:21:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3cd18c631b1e4dfc8da4a084e984cd35 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd66e6aff7cf4a37934495cbbae6131a : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d42f52543a434145921973c41b394cf4 : SUCCESS in 6m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/60eb5c635b8841bba889fdd4ac8e3e00 : SUCCESS in 19m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37f8586103444f4eb95fed9a281ae543 : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08bb5d308e574dec8f611017cd8b3dd0 : SUCCESS in 11m 51s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee5f3e329264403b8780c72f7ef59629 : SUCCESS in 27m 29s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bb17345edd346619c16b125370923ae : FAILURE in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9170ca7a3264b8413c07b2b0e3d00ad36898da8e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-04 20:51:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aac23a7d3ccb79119a4be540a641616f5aab9c7a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-05 00:36:19.000000000","message":"Patch Set 1:\n\nah.. right.. we\u0027ll get 7/10 up here asap","accounts_in_message":[],"_revision_number":1},{"id":"2e41bc110cabdc2530bf9a785d6ea3ef403614df","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-09 17:47:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"72166b12673a77f4860e66427b7c9da1a0d7f400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:17:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1cb66dcd8f04e31a919c1ee26277153 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21dc5ef1250f430daf8d2b0c6ae2b129 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57657aef96de4aeba8c87cc350ba9e17 : SUCCESS in 4m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/98866d174f134ea79cc34d00e8235e1a : SUCCESS in 19m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/eea5d827fd34413ba715f828f05acda7 : FAILURE in 9m 33s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/6eb25a08b6bd4ed0b979b3fb3d140c7f : FAILURE in 9m 03s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/816dcbe17af8488d98e1ade79779ef3d : SUCCESS in 11m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b62cfb4837e14f698452c6e627f49bd4 : SUCCESS in 11m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fc1e48a8f34b451495b03b3edf432e44 : SUCCESS in 26m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa715682d71a4398a80d9232dcd0be58 : FAILURE in 18m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a380bca74da2ba3be2b6acc471bf938d608100c5","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-10 13:00:41.000000000","message":"Patch Set 2:\n\nI do not see puppet-octavia-tripleo-standalone reporting, bad file triggers?","accounts_in_message":[],"_revision_number":2},{"id":"e1e7c960777ff0b145d957c01564abc294241951","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-10 13:11:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"70fe2100f8580de6333f9c14dbcc40124ad44a5a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-03-25 02:01:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5366008a7f0dc8acc875776d3671e1de91f7222","revisions":{"b11ce4963980c92de41c88473d58c853c767175e":{"kind":"REWORK","_number":1,"created":"2020-03-04 19:47:27.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/22/711322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/711322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/1"}}},"commit":{"parents":[{"commit":"81b45f9eab723a47265bb9d73725b2793a052167","subject":"Add Octavia OVN Provider configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81b45f9eab723a47265bb9d73725b2793a052167"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-03-04 19:46:15.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-03-04 19:47:26.000000000","tz":-420},"subject":"add tripleo job for octavia, centos-8","message":"add tripleo job for octavia, centos-8\n\npuppet-octavia has no tripleo job checking changes\nbut is allowed in via include_pkgs\n\nDepends-On: https://review.opendev.org/#/c/711320/\nChange-Id: Id1bd4753a876cd7c41461fb28789f7c6f8679b3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b11ce4963980c92de41c88473d58c853c767175e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b11ce4963980c92de41c88473d58c853c767175e"}]},"branch":"refs/heads/master"},"a5366008a7f0dc8acc875776d3671e1de91f7222":{"kind":"REWORK","_number":2,"created":"2020-03-09 17:47:49.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/22/711322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/22/711322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/22/711322/2"}}},"commit":{"parents":[{"commit":"81b45f9eab723a47265bb9d73725b2793a052167","subject":"Add Octavia OVN Provider configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/81b45f9eab723a47265bb9d73725b2793a052167"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-03-04 19:46:15.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-03-09 17:47:44.000000000","tz":-360},"subject":"add tripleo job for octavia, centos-8","message":"add tripleo job for octavia, centos-8\n\npuppet-octavia has no tripleo job checking changes\nbut is allowed in via include_pkgs\n\nChange-Id: Id1bd4753a876cd7c41461fb28789f7c6f8679b3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a5366008a7f0dc8acc875776d3671e1de91f7222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a5366008a7f0dc8acc875776d3671e1de91f7222"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
