)]}'
{"id":"openstack%2Fpuppet-octavia~711333","triplet_id":"openstack%2Fpuppet-octavia~master~Id74ef39ff303c9abf8118df203970cd871ed974a","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1861886","hashtags":[],"change_id":"Id74ef39ff303c9abf8118df203970cd871ed974a","subject":"Add Octavia OVN Provider configuration (1 of 2)","status":"MERGED","created":"2020-03-04 20:45:22.000000000","updated":"2020-03-25 09:01:15.000000000","submitted":"2020-03-20 22:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711333-1584745190709-577aa814","meta_rev_id":"405d483ba4a779ffbd37c1e42c36ab9c74b1aa3d","_number":711333,"virtual_id_number":711333,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 22:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-03-20 22:16:05.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-20 12:55:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2020-03-20 15:14:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-03-20 15:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":1,"date":"2020-03-20 22:16:05.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-12 21:27:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-12 21:28:21.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-03-13 08:37:25.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2020-03-18 10:53:27.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2020-03-20 12:55:00.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-20 15:14:55.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-03-20 15:22:25.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-20 22:16:05.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-20 22:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58a9fdbb1a06171776c83501573c9ea4efb444b2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-04 20:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a61623de3e9b9722b597a91b2bb50c01b2a7896f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-04 20:46:51.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2bf1661e020c6be00dce0fbb6c8c4306623d438b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-04 20:47:10.000000000","message":"Patch Set 2: Workflow-1\n\nNOT MERGING until we can ensure it will not break CI.","accounts_in_message":[],"_revision_number":2},{"id":"dd834f5d8f909e6139ad8ccc8327bbad407f75e7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-04 20:47:30.000000000","message":"Topic set to master (bug/1861886)","accounts_in_message":[],"_revision_number":2},{"id":"f7146ef5c53b274ba575434f25bb32cd96385937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 21:18:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5604283878e44f51bd4e244b3660aa5c : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fb4edeb729244f9b1661630b232cb5f : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16a8f5e5460d4423bec1541bcedd63a6 : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/09ece72121b84c48ae8f15b8078aa0e4 : SUCCESS in 20m 18s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f0200a12ceb4c0b81160427cdcf1d2e : SUCCESS in 11m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b6771d6182e414cbacfa85a88d2b6d6 : SUCCESS in 11m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/25b767042fb04bc186553bb68cfc16af : SUCCESS in 27m 21s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/951288bb3dcc45c6984a47f5b537d430 : FAILURE in 18m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9cb8aff45884b3db6aa27d99aa25551 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"da81d44e49b60b2c26ede27169dc91da463102b2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-05 15:18:35.000000000","message":"Topic changed from master (bug/1861886) to bug/1861886","accounts_in_message":[],"_revision_number":2},{"id":"1305fb560aa7aef459929fdf4306b81aa907c66d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-05 21:39:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5135daa2f23fee873bea133163d1c210fcb88ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 22:22:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b87139da0c54fccaf9bc7775afaeb15 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04e10d08ce53420c98c5bf91517d3d12 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe917ee8ec6b4d2daf5e27cf05e3f219 : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/18898d3528d242669b33b2aa132c9efb : SUCCESS in 18m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bc1864d9ee91419386e1f0c34d8f1022 : FAILURE in 5m 35s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/78ca3eb54b174794934582aeaeda1bf0 : FAILURE in 5m 33s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0322ca1a88d541bc92afb1b86937bce6 : SUCCESS in 10m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bab53e4b50f4613a7a1de033b5e1245 : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c2799d780ba6497ebaf1be5d66f5c3c7 : SUCCESS in 25m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d2badad22394042bd698b2221a13051 : FAILURE in 12m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d070ecf9504243bcb0bf475ae9870f3c : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"123a1774d022c37fd6fdc9f8aa73fc0b9c1bef18","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-12 21:35:39.000000000","message":"Patch Set 3:\n\nHello reviewers. I have created a temporary tht gerrit [1] to exercise this change to ensure we will not have a repeat issue of the last time [2]. :)  This is actually a partial if [3], so we can stage the changes in a way that nothing will break. Please lmk if this looks good to you.\n[1]: https://review.opendev.org/#/c/711921/  \u003d\u003d\u003e  ci/environments/scenario010-standalone.yaml\n[2]: https://review.opendev.org/#/c/711250/\n[3]: https://review.opendev.org/#/c/709618/","accounts_in_message":[],"_revision_number":3},{"id":"dbe5a8332f88c70209689fa38e96bcce31bff1b0","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2020-03-13 08:37:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6dbf9403c3c64f7a86721a61ef73bab771081ff5","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 18:39:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6ce7e4026f0c7a427dd340b720feae184d2ab15d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:11:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03bfe36efd384a98bef5ea029e7a9831 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37fff3780f8f4891a9e62b6591600508 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e18def8a9f24ae8b25f3d4c416449d0 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/111d1f04d31141f8937757394674b19e : SUCCESS in 18m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/92a94ff0304041369d264978b5b3ee04 : SUCCESS in 12m 32s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/3c72591f214546159e085602b6b6fa08 : SUCCESS in 12m 12s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb280db9d7e84a67b2de85e4d9dcd3be : SUCCESS in 12m 19s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40e254fbe7ff4d098ba3d325cc959442 : SUCCESS in 10m 37s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0927d1bdb5de466ba08db1fdb0fc3aac : SUCCESS in 26m 42s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/501c4d66b81d44579de5109dde9ddef9 : FAILURE in 12m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78cde022533a4a7e86b814de0368a180 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"9dec5c9cdd5eea8bf24999230597d00ab44b8fad","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-17 15:40:33.000000000","message":"Patch Set 3: Code-Review+2\n\ntested in tripleo via https://review.opendev.org/#/c/711921/","accounts_in_message":[],"_revision_number":3},{"id":"e6db338250fe6bb923a09ab17ea87ad5617d8221","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-17 16:35:46.000000000","message":"Patch Set 3:\n\nbut shouldn\u0027t 711921\u0027s tripleo-ci-centos-7-scenario010-ovn-provider-standalone have passed?","accounts_in_message":[],"_revision_number":3},{"id":"a4f2527faf1d9b4d0f0d365dfd4264f2d0dbde6f","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-17 17:35:10.000000000","message":"Patch Set 3:\n\n\u003e but shouldn\u0027t 711921\u0027s tripleo-ci-centos-7-scenario010-ovn-provider-standalone\n \u003e have passed?\n\nHmm... normally non-voting jobs are non-voting for a reason (too flaky?) and I do not pay much attention to them. Will recheck and see why this failed.","accounts_in_message":[],"_revision_number":3},{"id":"d9e11ab05f0b174cacf50bd1e4dab27af8cc4e7c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-17 17:35:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"a982f322bd0d28c26b7cfccc55d766693db30403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 18:07:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78080e8120b84a658677862ff7859039 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a7c646058274c78ad48f512e2c2d915 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85a480a071294c49954d90ecda079e4a : SUCCESS in 5m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/21aeca1bf4374c849cf8676a008fc00e : SUCCESS in 13m 08s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7103c23faf584251b86e05c7b7e401bf : SUCCESS in 15m 19s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/292d6d35714344d385fec370fd580dd1 : SUCCESS in 13m 20s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7e2cdaec28b4f08ba4e0e69f387794c : SUCCESS in 12m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcb51638d5894e68919f4dbd31d802f7 : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6f212b6a388d4099b3b179e00a154f12 : SUCCESS in 24m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69a87e44fbe84581a7ed73a5606a05b6 : FAILURE in 21m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955a920bebf243999ddf35a4c9d6240c : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"84dd37f4da923490469eabff8f8c95a1503057d6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-17 19:29:10.000000000","message":"Patch Set 3:\n\ni don\u0027t think the ovn job is passing yet? the previous issue was this broke the regular 10 job","accounts_in_message":[],"_revision_number":3},{"id":"eedd8bc3e7c9d9e580fd43878ab8df1080895144","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2020-03-18 10:53:27.000000000","message":"Patch Set 3: Code-Review+1\n\nsorry, I was confused by the scn10 job names","accounts_in_message":[],"_revision_number":3},{"id":"edfbd69a5e68ea132a120ab56c79f4a0f16560d7","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-18 13:19:57.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nthis isn\u0027t going to quite work the way we want","accounts_in_message":[],"_revision_number":3},{"id":"27aa7f35873d47269fa9a4df3a5ec82986dfffce","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-18 13:28:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66775f34cfd9c00c86e2587732b5b2ee0a89f273","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 14:00:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c51f527f1f521996d3de6a138b294627a50829b9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 14:00:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5fcbd950a6ea4bbb5b6503922bfe51dc83313f87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 14:32:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3853d88b05084b08aec7b9a3037e8d02 : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/181e2a8c7fe94eac8943f76bd22ac1ab : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7297d431f74e4e29bb2417f97d46379a : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/65249e0521874c9a9020453e86daf429 : SUCCESS in 19m 34s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/638eb41d42524432983b3bea742d59e1 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d633a02bf1dd44568dcf515a8112747c : SUCCESS in 12m 06s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e196d0ad1e004c299dfbfa2bf3d15783 : SUCCESS in 12m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ff587bf4c654477abe3a80b37fc0cf4 : SUCCESS in 10m 16s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1e6b5439597640369681a1c74b85d8b9 : SUCCESS in 26m 11s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6a5da8dcaa14568b1856f7007b34cc5 : FAILURE in 14m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d8e2115b871492a808dae0db693787a : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"14ddfb1fec28f016d0b0f8aa2c338270a6a2fe62","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 12:54:48.000000000","message":"Patch Set 4:\n\nI\u0027m not super into puppet but from what I can see it looks good to me. Thank you.","accounts_in_message":[],"_revision_number":4},{"id":"1b421a9bfb3da35698af5a1affbeb3e41786de84","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-20 12:55:00.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3f248e730d8e2c672dd4bec8724872e2633dfb03","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-20 15:14:55.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"153ff3351cf4af753fca44768224a9eca5209360","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-20 15:22:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"31ff85a35e72a4ab933f1ba34bf3da4b21cef528","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-20 22:16:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"193d3025df5428f7f6fdfb6ff4a469e713ae415d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:16:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"93ae0b56f67088f18a3bd930037c856244e5b7ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ba4e3954876acbe8c9a89658899fbd79e4a700c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 22:59:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb5db60708884c5e9f57d69e019e8064 : SUCCESS in 7m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c33cc3469e04fba9e0aadb4013dc283 : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a847258c36e24d8da76ed6ad4e17b7ed : SUCCESS in 6m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/71ed7612672847b19ec3fddc21b40e1c : SUCCESS in 13m 42s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddb31f00a7b5469cb19e69e791466467 : SUCCESS in 10m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bb78a6ba413044b89432bfaad1d915f7 : SUCCESS in 29m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cb73c423f9e4a76b009b0a598cba82a : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"3de5c3ffc3f2527f49704ce2faeef2319587b946","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 23:01:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7a7f5ed045b4866adfa84e1db5b3f48 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4},{"id":"405d483ba4a779ffbd37c1e42c36ab9c74b1aa3d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-25 09:01:15.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ecf7a8a32ac64dd6f6ff2310e7512952351e06e5","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dce599e49898d7973805a389d3abf5048f9f5099","revisions":{"9e3ec5380ef8c76dc745a67a75f999a054603a98":{"kind":"REWORK","_number":1,"created":"2020-03-04 20:45:22.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/711333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/1"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-04 20:45:03.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration","message":"Add Octavia OVN Provider configuration\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nRelated-Bug: #1861886\n(cherry picked from commit 81b45f9eab723a47265bb9d73725b2793a052167)\n\nChange-Id: Id74ef39ff303c9abf8118df203970cd871ed974a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e3ec5380ef8c76dc745a67a75f999a054603a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e3ec5380ef8c76dc745a67a75f999a054603a98"}]},"branch":"refs/heads/master"},"0931fa4ca12afeca745f9ec5eff9596a2c17606c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-04 20:46:51.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/711333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/2"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-04 20:46:17.000000000","tz":-300},"subject":"[DNM]: Add Octavia OVN Provider configuration","message":"[DNM]: Add Octavia OVN Provider configuration\n\nNOT MERGING until we can ensure it will not break CI.\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nRelated-Bug: #1861886\n(cherry picked from commit 81b45f9eab723a47265bb9d73725b2793a052167)\n\nChange-Id: Id74ef39ff303c9abf8118df203970cd871ed974a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0931fa4ca12afeca745f9ec5eff9596a2c17606c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0931fa4ca12afeca745f9ec5eff9596a2c17606c"}]},"branch":"refs/heads/master"},"3221095e95d58c8992623d00436e7066ed8fc6fe":{"kind":"REWORK","_number":3,"created":"2020-03-05 21:39:00.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/711333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/3"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:34:00.000000000","tz":-300},"subject":"Add Octavia OVN Provider configuration (1 of 2)","message":"Add Octavia OVN Provider configuration (1 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nThis is part 1 of 2, where octavia::api::ovn_nb_connection will move\nto newly created octavia::provider::ovn::ovn_nb_connection config file.\nBut that has to be split into 2 parts to avoid breaking the\nCI until the THT+pupple-tripleo changes merges [1].\n\n[1]: https://review.opendev.org/#/q/topic:bug/1861886+(status:open+OR+status:merged)\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nRelated-Bug: #1861886\nChange-Id: Id74ef39ff303c9abf8118df203970cd871ed974a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3221095e95d58c8992623d00436e7066ed8fc6fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3221095e95d58c8992623d00436e7066ed8fc6fe"}]},"branch":"refs/heads/master"},"dce599e49898d7973805a389d3abf5048f9f5099":{"kind":"REWORK","_number":4,"created":"2020-03-18 14:00:15.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/33/711333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711333/4"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-02-24 21:01:48.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-18 14:00:11.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (1 of 2)","message":"Add Octavia OVN Provider configuration (1 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nThis is part 1 of 2, where octavia::api::ovn_nb_connection will move\nto newly created octavia::provider::ovn::ovn_nb_connection config file.\nBut that has to be split into 2 parts to avoid breaking the\nCI until the THT+pupple-tripleo changes merges [1].\n\n[1]: https://review.opendev.org/#/q/topic:bug/1861886+(status:open+OR+status:merged)\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nRelated-Bug: #1861886\nChange-Id: Id74ef39ff303c9abf8118df203970cd871ed974a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
