)]}'
{"id":"openstack%2Fpuppet-octavia~711441","triplet_id":"openstack%2Fpuppet-octavia~master~Ic2dfa9b94458cf11241f77c0c7107617f6a2f138","project":"openstack/puppet-octavia","branch":"master","topic":"ut-controller","hashtags":[],"change_id":"Ic2dfa9b94458cf11241f77c0c7107617f6a2f138","subject":"Improve unit test coverage in octavia::controller","status":"MERGED","created":"2020-03-05 12:05:16.000000000","updated":"2020-03-10 00:31:59.000000000","submitted":"2020-03-10 00:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711441-1583800319939-a42b5d9c","meta_rev_id":"00eb145f9b6ab71fa9e1e75d641aaf1e68b98969","_number":711441,"virtual_id_number":711441,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 00:31:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-10 00:08:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-09 23:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-10 00:08:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 23:23:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-10 00:08:42.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-10 00:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f816814573fdc651a71a0c95e45db21955e86dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 12:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1932ec01bcc84e68abcfdf9cbe3399410cf1b1b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 12:07:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"898048ba6ca976094035b26e6d30f4f4960e265b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 12:07:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3093636c6b2496ce9affd352771ac4d219ba4630","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 12:09:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"90cad66f91d31ff8f58400a434e6c054ecaaaadf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:56:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94eed31531614d969b985a4d3d9c9938 : SUCCESS in 6m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b5fcbe84e484c71a2f822082ce3c0ef : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3da4e16fc3ab421f9a8cb4e63b061fb1 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/18179105e2724ff89dbcbaf30a145c88 : FAILURE in 17m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8f9942ae03e54bf5972c2c29f91ace2b : FAILURE in 15m 50s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/6f6dd25241904fb199954684eaeea675 : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe3daab846b040879f6b4fcfe73f89a7 : FAILURE in 10m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d564f869ff84705a69b8cb2cf8f0f04 : FAILURE in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35abe6c3ff52eee5edaaa20352d8b2dea86d879d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 14:01:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"30771777478c644bae81ff48418f82abfee4fe9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:27:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38e4e9f8402748e4a021c9b7b29d2bb7 : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed801147cf5b4f4787495fdc6dce61ee : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cea4500f33934ab58e1bbea870c98e71 : SUCCESS in 4m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1dba9c52f2d347a1b7a273a08b0d4172 : FAILURE in 21m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bf26e2f41ccf49d0bdf21788370005e2 : FAILURE in 7m 48s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/774ef3610bb1400999c46dc2a5000365 : FAILURE in 6m 52s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b5f1cf7490c4b6a88df94cb60fd1433 : FAILURE in 12m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa313e32f9524df087b3a794db54720f : FAILURE in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"325732930b6c25be1b9531b916c1616f235d6b78","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-05 14:42:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9fb440ac7cfc1e5e24b8d6a40e3ce0bcb7cc9d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 15:20:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd9f129a65dc4aacbe5ae1d1f734c967 : SUCCESS in 5m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75801c8952ca460e95c43703a1902f89 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2d85eeeea464fd19f50d899b6908104 : SUCCESS in 4m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1bea4119b4c9419181a2348400e5bf67 : SUCCESS in 18m 47s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/aff31011b76840cda9a21c71037e0161 : FAILURE in 8m 31s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2a583b3f19254b00a02bf9cf95e06e80 : FAILURE in 8m 13s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f94689d786da4db5b501befc3e7e3eaf : SUCCESS in 11m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10aa2cd7ce324d8e8444fe4123fadef7 : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a3bda35fdd16deeb3e300085561b89819fb646c2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 14:57:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a9cab62db52d9289ce32dc9e26fbce3fb72ec915","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-09 16:26:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3d687c64964dca811ea308120076d5a103344c28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:00:14.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"b1312f83be665ec409286c0cfd9bc9599e74c2c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-09 23:00:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"597f6e1e8f55c73fcb4647aeb36bee5428f2c312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 23:20:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15c39e3441e6464bb06b85ab3221c09d : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9263432bd2644e2cbf977dbd1bcd4e9b : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bf098c5c1a244f3939ebefa8b6683ce : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2c403fe5f7604676aa8e3dacda441a94 : SUCCESS in 17m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/41e49ff55f6b4be3a8ce2c86ef7d8992 : FAILURE in 6m 23s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7058be1e74774effa118c5eb37bbe786 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b62e4957e1d84e0dbb693c44f6dd4be4 : SUCCESS in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5464ed7a63f947c892aaeeab43aee5bb : SUCCESS in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"462b653c6286118690a35b27986de9f3d858fd48","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-09 23:23:48.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"fcb34d6176f2d329fd02dfeac3a85f97a8756e16","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-10 00:08:42.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a2fc4c5b4493a5d4ce12d4354cdbd8fcd0243ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:08:57.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"789ddd3d961bf5869a18200f095ec6dc6f72f0d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:31:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"cc3184dfff056a7805534133672d07607555b4a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 00:31:59.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6524a5bc31044af6af4a2d877c3af8fa : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68e47c76a2c8435fbe32829f3172edf9 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af4064fad6fd44bd88f54a9065f0c370 : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d4f1b358780f4182919786587f9f66b2 : SUCCESS in 18m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/491e16c296534d9883599c8f1caf7dee : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c67e6fa14263a0a2201646678c663b880c6c6cd6","revisions":{"991ddf2f2816f2a5d53add0d8d8b4c54b4001c36":{"kind":"REWORK","_number":1,"created":"2020-03-05 12:05:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/1"}}},"commit":{"parents":[{"commit":"67501dfe117afb3842a83f6b11296b7c076edb85","subject":"Support amp_image_owner_id option for octavia controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/67501dfe117afb3842a83f6b11296b7c076edb85"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:04:54.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/991ddf2f2816f2a5d53add0d8d8b4c54b4001c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/991ddf2f2816f2a5d53add0d8d8b4c54b4001c36"}]},"branch":"refs/heads/master"},"06c71e60d2c08ba19709c3dfadfa0d1cafce18ac":{"kind":"REWORK","_number":2,"created":"2020-03-05 12:07:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/2"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:06:56.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/06c71e60d2c08ba19709c3dfadfa0d1cafce18ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/06c71e60d2c08ba19709c3dfadfa0d1cafce18ac"}]},"branch":"refs/heads/master"},"2053c3f2fa30f66aad2d46e2851b20fb08772922":{"kind":"REWORK","_number":3,"created":"2020-03-05 12:07:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/3"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:07:47.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2053c3f2fa30f66aad2d46e2851b20fb08772922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2053c3f2fa30f66aad2d46e2851b20fb08772922"}]},"branch":"refs/heads/master"},"c9b406b416257c49c78a984507d590c10f115179":{"kind":"REWORK","_number":4,"created":"2020-03-05 12:09:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/4"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:08:57.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c9b406b416257c49c78a984507d590c10f115179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c9b406b416257c49c78a984507d590c10f115179"}]},"branch":"refs/heads/master"},"9dc5f04f1a55160ea40e7ad3125fa3aef8a0af76":{"kind":"REWORK","_number":5,"created":"2020-03-05 14:01:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/5"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 14:01:01.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9dc5f04f1a55160ea40e7ad3125fa3aef8a0af76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9dc5f04f1a55160ea40e7ad3125fa3aef8a0af76"}]},"branch":"refs/heads/master"},"57eda4a2e0d4359a5eadd7470035c450fce1d146":{"kind":"REWORK","_number":6,"created":"2020-03-05 14:42:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/6"}}},"commit":{"parents":[{"commit":"a3c06a4c651b14c2cae197faee20377f8aa7433e","subject":"Merge \"[ussuri][goal] Cleanup for python 2.7 drop\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3c06a4c651b14c2cae197faee20377f8aa7433e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 14:42:54.000000000","tz":0},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/57eda4a2e0d4359a5eadd7470035c450fce1d146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/57eda4a2e0d4359a5eadd7470035c450fce1d146"}]},"branch":"refs/heads/master"},"c67e6fa14263a0a2201646678c663b880c6c6cd6":{"kind":"REWORK","_number":7,"created":"2020-03-09 23:00:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/711441/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/41/711441/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/41/711441/7"}}},"commit":{"parents":[{"commit":"2afe375f788facd6c59d9389a370452a1fc938ef","subject":"Merge \"Adds amp_active_retries and amp_active_wait_sec\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2afe375f788facd6c59d9389a370452a1fc938ef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-05 12:03:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-03-09 23:00:32.000000000","tz":540},"subject":"Improve unit test coverage in octavia::controller","message":"Improve unit test coverage in octavia::controller\n\nAdd missing tests for octavia::controller so that we can test all\nparameters with their default values and overridden values.\n\nChange-Id: Ic2dfa9b94458cf11241f77c0c7107617f6a2f138\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c67e6fa14263a0a2201646678c663b880c6c6cd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c67e6fa14263a0a2201646678c663b880c6c6cd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
