)]}'
{"id":"openstack%2Fpuppet-octavia~711557","triplet_id":"openstack%2Fpuppet-octavia~master~Ic2602e7a219f3c3be7f085726ee9736975062306","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1861886","hashtags":[],"change_id":"Ic2602e7a219f3c3be7f085726ee9736975062306","subject":"Add Octavia OVN Provider configuration (2 of 2)","status":"MERGED","created":"2020-03-05 21:42:50.000000000","updated":"2020-03-31 16:23:33.000000000","submitted":"2020-03-31 16:21:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711557-1585671678259-eecf9550","meta_rev_id":"a3a8899309c82f27ea485bef545e64ab7e5209dc","_number":711557,"virtual_id_number":711557,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 16:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-03-30 22:13:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-30 13:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-03-31 15:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-03-30 17:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-30 23:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-03-31 15:40:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:49:35.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-03-30 17:14:05.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-03-30 22:13:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-30 23:14:13.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2020-03-31 15:40:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-31 16:21:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2befe7e9a94fd359672211e2658beeb20efbbb26","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-05 21:42:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a342d5739de52d327768028a35a03b41ed7d4854","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-05 21:43:20.000000000","message":"Topic set to bug/1861886","accounts_in_message":[],"_revision_number":1},{"id":"b3e01f3c8570363b65a6abce1be0d6306a8949ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 22:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/712181c87453408d940415b86507d318 : SUCCESS in 7m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4d8f220ad3a4195af4f162a4b7e4c80 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff18a3238ac94fdc8dea747d80ae3284 : SUCCESS in 5m 11s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b0ac1fd5c71f43309b02059f30bc21d2 : SUCCESS in 15m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/090c31231f594e01a87b86dd8ca1aa5c : FAILURE in 7m 47s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/fe81581c226d4bef956edc0208d4af7f : FAILURE in 10m 05s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13d66fb8dd504230a66bcb4474e40573 : SUCCESS in 11m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b80fcf7c72849e692f5dcda2f9c17ff : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/29da8bf22dcc4ab3a0bb45178eeca572 : SUCCESS in 23m 25s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a038398fb6da48f9abdeee2736598247 : FAILURE in 17m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14920510a7d0464b9d1ad1d8d6ce994f : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a6f1b6f2a92c4de9b498944072f3277a11d0fa3e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 18:51:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"749e5a634b67508ea1b3bfb2c6d05cab25471f1c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 18:59:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b9b1238f503546c591991885471b4c3e1fd6c21","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 19:00:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"abd8b5b777812d50b6fc07604ba5fc99b72b0685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 19:34:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27b4c0bab8964f25abba9a4a683acbb4 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/708422185fb94e1080edd8400c14f6d1 : SUCCESS in 6m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97dea7e5d57c4f179074827be8a6d710 : SUCCESS in 6m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/94d1ab6fa99f40479ffaf553a8032e01 : SUCCESS in 17m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c9c60d0aa5294d039bb8497d5f5a65ef : SUCCESS in 16m 50s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0fdb97b8bc164b2ab06aa6e582a69fbd : SUCCESS in 11m 47s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f157d1cadf0e4c5f9e778e324ac0127b : SUCCESS in 10m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19f9963fcba04182b5fb13103dc456b7 : SUCCESS in 17m 18s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d97dba0a3c7641a0830d088ad64239f9 : SUCCESS in 28m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d5b10db54fe48d8a078a24a9fda14f4 : FAILURE in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd73eb7ab4274b05b2776043440664b8 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ff5169f1719ac473ed01605dd62096c88e2b786e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 21:42:39.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"be1e10906b4609152831c9c910b61cf720806a1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 22:18:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/032f1c4031954befb0b3874a9d6dc19a : SUCCESS in 5m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4060c9fdded04b7f8c69996f1a24a14d : SUCCESS in 6m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f735a8abba07482789de76c33d3d011e : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c9854e872dda48278fd3900de1427aea : SUCCESS in 14m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/055c37bb591e4a0bb3b9ad659d4e757e : SUCCESS in 14m 48s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/04f630dac49943a0b7456ef5210ab8f2 : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a18da7578c0148909b8e6ce7a6a4d63e : SUCCESS in 13m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2434519f492b42bdaafbfc806b0945c7 : SUCCESS in 13m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9dc5bc9d8182492fa131e5e97850e410 : SUCCESS in 24m 19s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7563c935990c4f558e8a0a68995ccccc : FAILURE in 12m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8523826fa3984dc08be4af1c9bcc41db : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"90715070652758f42e749ed7101dc6159a58c274","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-20 20:40:36.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"735701dcdd9900b75c03f2355a99a2508f2a0e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 21:10:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/005cb6392f2e489bb116ac8c861d91bc : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/042e0aca5ef44460a289fa2c2dbffe59 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/054125b3c5ca44e2aa228a32f6f04633 : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/df54ba187cee4fddb8580356b5924b63 : SUCCESS in 16m 58s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bf197731f1f14bda8df7c4def3dee2be : SUCCESS in 14m 07s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a7f3912dbf1e45a694f314a70dab3185 : SUCCESS in 14m 05s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cfe3a0571a64ddca4e5917d9fcdfcb3 : SUCCESS in 10m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c60762cadfb4ab1b9b477659b38c896 : SUCCESS in 11m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/968d809d20dd426bbfbe2e2195f3de9c : SUCCESS in 27m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ee16ea6f3a649c2a366d83eb02070d6 : FAILURE in 7m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db3afc71ff2545a9a0a5e49c1b82fc2b : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":5},{"id":"2975abd8969f2157ad7e60adc81f07c2c2490904","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-23 23:19:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"5dd8a9137a436af26b9e6dbff37817cab5abdee2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-27 20:51:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1166c09c67f9a35cccea448201eeccfdf2f00efe","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-27 21:08:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0edfd6367feb9c9f64bef5f9316bafffa9850da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 21:21:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e720b251221f42748cff3764146a8cb2 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fc48ce5e8be400095d1f98cd4cf78a4 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba600975b3684144bd853cf27ed235a5 : SUCCESS in 5m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/efc9b34e50a741a9bc0ee728c610005a : SUCCESS in 21m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7ac0d274d49a42739ceff0c5281baf54 : SUCCESS in 10m 29s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a2ba5c70f7ed40c59880dcbfb5ce465b : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/883d13d6e34f4462828b598d029ae736 : SUCCESS in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38186c9e78d7453c930073c3424f772a : SUCCESS in 9m 40s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/358eefb704134eeebb6cba9cd3e6d6ee : SUCCESS in 24m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2fa5ff917ab4098a169f0106bf0e128 : FAILURE in 6m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4625f7258d34af3833096375d4a85c0 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":6},{"id":"568d80d9f7700c559cedf9753a3443412e1e0554","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-30 12:10:51.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f14fcc72472491497722c51e388a1cf07292a8b1","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-30 12:14:39.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)\n\nLGTM - maybe add comment as suggested if you respin?","accounts_in_message":[],"_revision_number":6},{"id":"d6941249c8a0bb78d2ab519e47464d7169f30556","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-30 13:17:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9ac66633c4a633013eb8836c7958ab84e4493aed","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-30 13:19:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a7891075331f510de6a8f5424cf291dde8d8479d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-30 13:34:34.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9114d9cc32ff4c8c2664e80a9bd5e5661e91ca71","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-30 13:41:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ebf0c59a0cd4f11439656dc3951fb9837450d73c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-30 13:42:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5c0e8e3c731219c24d09cf5f7097c6c17901d90d","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-30 13:49:35.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0c5cb6152ccf090371e2e34244dffb04e7522280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 14:35:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0181b6fc4ce4113b4d68ff68109b75d : SUCCESS in 6m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f545f8b04bb4166830e0b21e97a12cf : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/308b7a34ea9448e99b61014744302e27 : SUCCESS in 6m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bc951d5e8e29490081df0f36bad70c77 : SUCCESS in 21m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/086027ac93b84009850ecba1ce3eb679 : SUCCESS in 13m 03s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/fbb30d3b8cf941eeae5df84fdc55f431 : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23704121d4d442b7a5f895106be7c70d : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a1e81d422e040c08e3df17ac28b283d : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7e055896fc83434e8d0fb78ac760e567 : SUCCESS in 30m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c1c850390c46e88bbddbdeb4cc8821 : FAILURE in 9m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e5c202159724401872be09e2747b90d : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":8},{"id":"3ee9ff5a0a7d86127cea6a4842b1899bcbe2fa65","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2020-03-30 17:14:05.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e3dc0723928ca4c6848cf39d254d754e0342db06","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 22:13:03.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"8e8b4ee3ef00eb31548bc5f8ac24725f90097abb","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2020-03-30 23:14:13.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"d57028ddfd4e54e8e4e7749d819a3abc24d1d2ab","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 15:40:09.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1067a271a01cf98654542fd8835296442fd944fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:40:21.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dd651b33c5d1c2c91aeb613700c70361bc2b6658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:21:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cc9b4f3601e4f13a3edeb5a4b6f8ffc : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e1b98585ac546319c3b3d542fc32f08 : SUCCESS in 7m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4b167d1e18b4df8941ef9d352e69339 : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/198d0ab8207b48c0aed1f9fe26d102c2 : SUCCESS in 15m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a90ec46171c744a9ad17cb904bbc4410 : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d7f10aebd3924fc1be16b12ce4ae6f87 : SUCCESS in 11m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c9743b3eff04393ad93c143c1952e24 : SUCCESS in 14m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/de337fb3f238485e8c2d0ea0dc4b7830 : SUCCESS in 27m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/defb526931f74a2b96f38b0dc0e40895 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":8},{"id":"ab4c7e415cabb1f84e311e69f1b88113da24008c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:21:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"a3a8899309c82f27ea485bef545e64ab7e5209dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:23:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd3a50585b744af68f01e199a301735b : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"8777b5031c611b7a4dd9cc9afbf0e65a0069ccf1","revisions":{"8bc9eca333b6bf9dcfc0a8698fcb9fc8e5013903":{"kind":"REWORK","_number":1,"created":"2020-03-05 21:42:50.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/1"}}},"commit":{"parents":[{"commit":"3221095e95d58c8992623d00436e7066ed8fc6fe","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3221095e95d58c8992623d00436e7066ed8fc6fe"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"subject":"[DNM]: Add Octavia OVN Provider configuration (2 of 2)","message":"[DNM]: Add Octavia OVN Provider configuration (2 of 2)\n\nDO NOT MERGE before changes to THT+puppet-tripleo for using\nnew config file is merged.\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nCo-Authored-By: Brent Eagles \u003cbeagles@redhat.com\u003e\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8bc9eca333b6bf9dcfc0a8698fcb9fc8e5013903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8bc9eca333b6bf9dcfc0a8698fcb9fc8e5013903"}]},"branch":"refs/heads/master"},"4e172169610a9c19960014bc5811acac8bd91c44":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-18 18:59:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/2"}}},"commit":{"parents":[{"commit":"3221095e95d58c8992623d00436e7066ed8fc6fe","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3221095e95d58c8992623d00436e7066ed8fc6fe"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-18 18:59:16.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nDO NOT MERGE before changes to THT+puppet-tripleo for using\nnew config file is merged.\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4e172169610a9c19960014bc5811acac8bd91c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4e172169610a9c19960014bc5811acac8bd91c44"}]},"branch":"refs/heads/master"},"651a18fe8b1c6b3f1257932c87dfddebdd5fe768":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-18 19:00:18.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/3"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-03-18 19:00:18.000000000","tz":0},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nDO NOT MERGE before changes to THT+puppet-tripleo for using\nnew config file is merged.\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/651a18fe8b1c6b3f1257932c87dfddebdd5fe768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/651a18fe8b1c6b3f1257932c87dfddebdd5fe768"}]},"branch":"refs/heads/master"},"f0d4e273c5d8efc54382709d8855e7ee6153a22d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-03-18 21:42:39.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/4"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-18 21:42:25.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nDO NOT MERGE before changes to THT+puppet-tripleo for using\nnew config file is merged.\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nDepends-On: https://review.opendev.org/#/c/707695/\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f0d4e273c5d8efc54382709d8855e7ee6153a22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f0d4e273c5d8efc54382709d8855e7ee6153a22d"}]},"branch":"refs/heads/master"},"764c2ad99ff75ac25db86f951f2d265afb6bd18b":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-03-20 20:40:36.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/5"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-20 20:40:16.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nDepends-On: https://review.opendev.org/#/c/707695/\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/764c2ad99ff75ac25db86f951f2d265afb6bd18b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/764c2ad99ff75ac25db86f951f2d265afb6bd18b"}]},"branch":"refs/heads/master"},"d74aaf551ecbd21bf3b7c03f2aa926a66a9991a4":{"kind":"REWORK","_number":6,"created":"2020-03-27 20:51:32.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/6"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-27 20:26:25.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d74aaf551ecbd21bf3b7c03f2aa926a66a9991a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d74aaf551ecbd21bf3b7c03f2aa926a66a9991a4"}]},"branch":"refs/heads/master"},"7d01a984fc27c7420227684fb21b27d491c21c59":{"kind":"REWORK","_number":7,"created":"2020-03-30 13:17:05.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/7"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-30 13:17:01.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7d01a984fc27c7420227684fb21b27d491c21c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7d01a984fc27c7420227684fb21b27d491c21c59"}]},"branch":"refs/heads/master"},"8777b5031c611b7a4dd9cc9afbf0e65a0069ccf1":{"kind":"REWORK","_number":8,"created":"2020-03-30 13:41:28.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/57/711557/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/711557/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/711557/8"}}},"commit":{"parents":[{"commit":"dce599e49898d7973805a389d3abf5048f9f5099","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/dce599e49898d7973805a389d3abf5048f9f5099"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-05 21:40:44.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-30 13:41:24.000000000","tz":-240},"subject":"Add Octavia OVN Provider configuration (2 of 2)","message":"Add Octavia OVN Provider configuration (2 of 2)\n\nThis patch enhances Octavia\u0027s OVN driver config, so it can connect to\nOVN_Northbound DB using TLS.\n\nRelated-Bug: #1861886\n\nChange-Id: Ic2602e7a219f3c3be7f085726ee9736975062306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8777b5031c611b7a4dd9cc9afbf0e65a0069ccf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8777b5031c611b7a4dd9cc9afbf0e65a0069ccf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
