)]}'
{"id":"openstack%2Fpuppet-octavia~711733","triplet_id":"openstack%2Fpuppet-octavia~master~Ie421204d7de3d9da9c4c3c87fea6ac8e9c292aa9","project":"openstack/puppet-octavia","branch":"master","hashtags":[],"change_id":"Ie421204d7de3d9da9c4c3c87fea6ac8e9c292aa9","subject":"Adds amp_active_retries and amp_active_wait_sec","status":"MERGED","created":"2020-03-06 18:20:06.000000000","updated":"2020-03-09 17:04:55.000000000","submitted":"2020-03-09 17:00:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711733-1583773214317-72e87165","meta_rev_id":"ea6eaf00b7bcf9ae7cf656ad9dadc85332e1b81a","_number":711733,"virtual_id_number":711733,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-09 17:00:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-09 14:57:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-09 16:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-09 16:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 14:57:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-09 16:24:39.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-09 17:00:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff1dd5cb3e9563429f82ab0984be98ab99f42277","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-06 18:20:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f004449750d4ddf4f0d73bb96d60b2b27eb6364","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:23:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22cc67ef6591463bafd397a1ed9fe3eb : SUCCESS in 5m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ecfcf773abf4aa49a1ea39cd5565d4e : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/597ef2923b61490280e3c2ffa6596039 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b888ba4022854d2cb40a242901539ac9 : FAILURE in 20m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9fdf8bd03b85429db07e0e56e8097f01 : FAILURE in 10m 14s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/ce597ba9646e473b805024f015f54ac3 : FAILURE in 8m 16s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/464a1f5d920043e6a7f0eabb70eae393 : FAILURE in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7f83ed36f224eafbbca8daa6999d1ce : FAILURE in 10m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aafd31d91aac45b8b6037fd983816c91 : SUCCESS in 27m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc9f31dfcfb1487aab06a82ca7cd292c : FAILURE in 14m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d9a49d887ae4521bd36042334b01186 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"67dbc1bc6c989eac5eff7184ecd38a7cf7c23623","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-06 20:43:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9350a7ff6d1f0071313cb1a1e52e0582b2e5a2ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 21:14:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6687986d21df4b099c295ffceccfada1 : SUCCESS in 5m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49dcd63bc69d4673a2d86a9872ab2855 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32fc820380fe44ad990bdfaae184a998 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/195e9551d68445a1a667c46fb5af2ef6 : SUCCESS in 18m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/751f4887b7ed4ec598c250c9f9ee270d : FAILURE in 6m 34s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c5b22c66a0ae4ab4a1a57975e296c538 : FAILURE in 11m 24s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a7d0244783a4fe89431fa1543eaa6fd : SUCCESS in 8m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/428eec4d8cbc417cb322bcc6c2ddef4f : SUCCESS in 10m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d7d962d6f7cd47f287562e40cf6abc23 : SUCCESS in 24m 34s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fd0b9aeb4194fb4a408aba6bab26e83 : FAILURE in 22m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70175e328e7a430d98cdbc07354aec0e : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"da7cdf1d19a892c31b53425599122f2a9746e10f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-09 14:57:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7b3b8fabb60976ddcca60351d3bfa01b6d8611cb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-09 16:24:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"16d62a0d8eae9f3088d12b59f77c353490623a02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:24:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56d3029b6c0f455dfbde60737b2920f04ed11aa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:00:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bede705d8ecf4ba1874fd388566a4adf : SUCCESS in 7m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66049b90d39d450ba2ecac3086e5cfa1 : SUCCESS in 7m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16d810dde29c40b4b6320207c84063f2 : SUCCESS in 5m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f6817298c06c49c48e54fd74d82f012d : SUCCESS in 21m 11s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89b901003d96443c93672073d572de9d : SUCCESS in 13m 04s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9884d5dd594646e382979d4f267d2f5b : SUCCESS in 32m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2a070a131af4976937e3102e474796b : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"689433e66018c921f24d517421d4aa2d5edea589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:00:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea6eaf00b7bcf9ae7cf656ad9dadc85332e1b81a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 17:04:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a565f804f644aa88cf56f7ac9902f73 : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f652397169f165c02adb904008663a4cf4fbb5f","revisions":{"331ca9571c1299d73482e6ebda1caa19d03fab6c":{"kind":"REWORK","_number":1,"created":"2020-03-06 18:20:06.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/33/711733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/1"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-06 18:12:02.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-06 18:12:02.000000000","tz":60},"subject":"Adds amp_active_retries and amp_active_wait_sec","message":"Adds amp_active_retries and amp_active_wait_sec\n\nChange-Id: Ie421204d7de3d9da9c4c3c87fea6ac8e9c292aa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/331ca9571c1299d73482e6ebda1caa19d03fab6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/331ca9571c1299d73482e6ebda1caa19d03fab6c"}]},"branch":"refs/heads/master"},"3f652397169f165c02adb904008663a4cf4fbb5f":{"kind":"REWORK","_number":2,"created":"2020-03-06 20:43:52.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/33/711733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/33/711733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/33/711733/2"}}},"commit":{"parents":[{"commit":"b31859d8eecaff8c218c355e59f1a0292681ab27","subject":"Revert \"Add Octavia OVN Provider configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b31859d8eecaff8c218c355e59f1a0292681ab27"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-06 18:12:02.000000000","tz":60},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-03-06 20:43:41.000000000","tz":60},"subject":"Adds amp_active_retries and amp_active_wait_sec","message":"Adds amp_active_retries and amp_active_wait_sec\n\nChange-Id: Ie421204d7de3d9da9c4c3c87fea6ac8e9c292aa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f652397169f165c02adb904008663a4cf4fbb5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f652397169f165c02adb904008663a4cf4fbb5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
