)]}'
{"id":"openstack%2Fpuppet-octavia~717091","triplet_id":"openstack%2Fpuppet-octavia~stable%2Ftrain~Ica6f84256ff2068e9bac210e66a1429ea627256c","project":"openstack/puppet-octavia","branch":"stable/train","hashtags":[],"change_id":"Ica6f84256ff2068e9bac210e66a1429ea627256c","subject":"Add driver agent configuration","status":"MERGED","created":"2020-04-02 18:13:57.000000000","updated":"2020-04-03 14:51:07.000000000","submitted":"2020-04-03 14:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717091-1585925310718-176f1ad0","meta_rev_id":"4c552e3dc73ff22d676f2adb22147e1bde855422","_number":717091,"virtual_id_number":717091,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 14:48:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-03 14:14:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2020-04-03 00:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-03 14:14:40.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 18:13:57.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2020-04-02 18:44:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-03 00:30:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-03 14:14:40.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-03 14:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f949eef6f512f91b7d55ec521f0710d87665b8e0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 18:13:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"62fafb3cca5f21372ca2715007f379d7c97d11e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:56:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f41294b135704edb8425b9b530d77670 : FAILURE in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8e317a4f3524c1387e8f167a74c9c19 : SUCCESS in 7m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dede83b8d6bf41f2b97f41074ee18bed : SUCCESS in 7m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bafa07f8c5e84c39a20d0bf0194d8fff : SUCCESS in 15m 12s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7e64e642cc2f4022bc25045badbae56c : SUCCESS in 15m 42s (non-voting)\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/4ab177a4e7e94443a655ae6ff950c40e : SUCCESS in 15m 24s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9c324c2f77324745ab3357b61f2631b6 : SUCCESS in 32m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/492a0228a54f4ad9adc15ccc5c00f612 : FAILURE in 6m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a18934709f4466b941427fcc30c84d0 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ec376004ffa74de4a23ddc16d17e3d3feeb887bf","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-02 19:23:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0ba780ba287f1ae1f556b7773623e2542c828005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:58:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa5f99bd6d85473ca1a85fce77377809 : SUCCESS in 5m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4c10032866f4e9bbd925eda5bd87719 : SUCCESS in 6m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab6e4d9a720742c98646a2b58827e4cf : SUCCESS in 7m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7d6b457f0b834a50be09c81a4db67bf2 : SUCCESS in 19m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/28f6fdf3f5c34228bd7472aeb7150c65 : SUCCESS in 16m 34s (non-voting)\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/09db22829b2f48e4adaf9f35c8a62df1 : SUCCESS in 18m 02s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3fd0d1fb74a643f2b02b45c66d42cbf9 : SUCCESS in 25m 41s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfb56cd5bfa34994adfb3228f73be05b : FAILURE in 7m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8361ba6263e34e92859c19f0304adf3c : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"27ab075545d0c7f0727effe8d51e8e6958a2b69b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-03 00:30:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"416000e0e2f8eeb769f072a353e4ff09ab7dce28","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-03 14:14:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8472983121d715abd6ed096fa8ea8d5aff6f9972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:14:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"343d65272fde40c54153d77f64bd0db0ce2368c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:48:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"94206c2c34d4874ddf9e81a29f847a2ae021581f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:48:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ca2d7a2d76b4ae8a48c262a35d328ed : SUCCESS in 5m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad7194d6e5684e5ca133eb413943be4a : SUCCESS in 6m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48aa0d4a5b79427596d1278beee7d760 : SUCCESS in 6m 57s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7753eab6a81b42a9a679843d4c812cbc : SUCCESS in 21m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e5110b56810f46b08dbcadb1862bd3d0 : SUCCESS in 25m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30dc3bb8ebdd43c2b14607e699f4f697 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4c552e3dc73ff22d676f2adb22147e1bde855422","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:51:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4907f06678da497bb39623392a2df33d : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6284d53eb78a3fdac9ab6b1a5be362b8cd7adaee","revisions":{"397d5fb86384fa59f225ca367d15207c40965912":{"kind":"REWORK","_number":1,"created":"2020-04-02 18:13:57.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/717091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/91/717091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/1"}}},"commit":{"parents":[{"commit":"a3603e6161d8d6a712131dfa8022ddc11e7d5557","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3603e6161d8d6a712131dfa8022ddc11e7d5557"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-02 18:13:57.000000000","tz":0},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n(cherry picked from commit 88ff210c3021583921f41532b190fe0c3bafca62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/397d5fb86384fa59f225ca367d15207c40965912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/397d5fb86384fa59f225ca367d15207c40965912"}]},"branch":"refs/heads/stable/train"},"6284d53eb78a3fdac9ab6b1a5be362b8cd7adaee":{"kind":"REWORK","_number":2,"created":"2020-04-02 19:23:46.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/717091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/91/717091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/91/717091/2"}}},"commit":{"parents":[{"commit":"a3603e6161d8d6a712131dfa8022ddc11e7d5557","subject":"Add Octavia OVN Provider configuration (1 of 2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a3603e6161d8d6a712131dfa8022ddc11e7d5557"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2019-06-05 18:09:58.000000000","tz":0},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-04-02 19:23:43.000000000","tz":0},"subject":"Add driver agent configuration","message":"Add driver agent configuration\n\nAdd support for octavia driver agent configuration options.\n\nChange-Id: Ica6f84256ff2068e9bac210e66a1429ea627256c\n(cherry picked from commit 88ff210c3021583921f41532b190fe0c3bafca62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6284d53eb78a3fdac9ab6b1a5be362b8cd7adaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6284d53eb78a3fdac9ab6b1a5be362b8cd7adaee"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
