)]}'
{"id":"openstack%2Fpuppet-octavia~724559","triplet_id":"openstack%2Fpuppet-octavia~master~Iff5220af55614fc7c5599b39304906ff7a4f0ce9","project":"openstack/puppet-octavia","branch":"master","topic":"bump-openstackdocstheme","hashtags":[],"change_id":"Iff5220af55614fc7c5599b39304906ff7a4f0ce9","subject":"Bump openstackdocstheme to 1.30.0","status":"MERGED","created":"2020-04-30 07:29:23.000000000","updated":"2020-04-30 08:07:59.000000000","submitted":"2020-04-30 08:05:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724559-1588233946928-2ccb4115","meta_rev_id":"d0f4b47508efdd3f13e10f44fa192d61ccaf0a73","_number":724559,"virtual_id_number":724559,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 08:05:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-30 07:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-30 07:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 08:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4281878a10c7941a760d0417ca94092356cf6b4a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:29:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d995b3c20e849120b9d0b3f0bc419344c10356f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34a1c59c40104435b087a3b1862fd1bc : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5235a08198372a903f282681d478317f265572ea","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"471058d387d9b10b9bf91f9cfdbf61685d688786","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:57:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdfbe4d74583a5659fb0aaa256ece5a0400a4f74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:05:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c8fbbac440045879d9cc5665c440b49 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d4585356e638ae50d9747bc80a70dc6a648f94bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:05:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"36d0581b66c628233fe78410b73612612f61bf8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:07:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98872ec861bb4caab3d93a839ea96c0a : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64778fccd06fa82b3af78a30eb87dad3c3aaf955","revisions":{"64778fccd06fa82b3af78a30eb87dad3c3aaf955":{"kind":"REWORK","_number":1,"created":"2020-04-30 07:29:23.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/59/724559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/59/724559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/59/724559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/59/724559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/59/724559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/59/724559/1"}}},"commit":{"parents":[{"commit":"95927665ea41dd0cdc19fead5a664449e26be992","subject":"Fix rabbitmq version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/95927665ea41dd0cdc19fead5a664449e26be992"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.30.0","message":"Bump openstackdocstheme to 1.30.0\n\n...to pick up many improvements, including the return of table borders.\n\nChange-Id: Iff5220af55614fc7c5599b39304906ff7a4f0ce9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64778fccd06fa82b3af78a30eb87dad3c3aaf955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/64778fccd06fa82b3af78a30eb87dad3c3aaf955"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
