)]}'
{"id":"openstack%2Fpuppet-octavia~740942","triplet_id":"openstack%2Fpuppet-octavia~master~I18e2f3376fefea600a1ad5a931ce52310b44f51f","project":"openstack/puppet-octavia","branch":"master","topic":"at-interface","hashtags":[],"change_id":"I18e2f3376fefea600a1ad5a931ce52310b44f51f","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-07-14 11:28:55.000000000","updated":"2020-07-14 18:32:28.000000000","submitted":"2020-07-14 18:30:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740942-1594751443647-ffcc4a9e","meta_rev_id":"9731bebdf4e9fb43f7dccb0200716f811b3a055f","_number":740942,"virtual_id_number":740942,"owner":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-14 18:30:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-14 18:00:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-07-14 16:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-07-14 18:00:51.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 16:41:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-14 18:00:51.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-14 18:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"32faadd55c52381eaf10a156766b9243efcd3317","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2020-07-14 11:28:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e0794756a729e929ab54b7c79cc69041ea03c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 12:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4fd6d286fbf46db92da45f48e12832e : SUCCESS in 5m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24ed5ef64a8145d8818162edbfab4461 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f2deace0e0445e18354aaceb68e15d6 : SUCCESS in 4m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8b2cb3e8a5df4d67b1aa4ca8d2763c94 : SUCCESS in 14m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bb1b9864f94f43c49be24029ee61352a : SUCCESS in 15m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dc77da21fc3442fadf99ff1600cddec : SUCCESS in 10m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f08a48221641467485f9add0479c674a : SUCCESS in 13m 44s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/105ce2db71454fab85e6c8a7e67ee7b9 : SUCCESS in 26m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c0aea00985e14338928061aecee77fd0 : FAILURE in 8m 28s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07617ce0b65a4cb9b436a3928fc6206c : FAILURE in 6m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/408d291deb034cccbb65f4e39bf7a413 : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6abbdab5d5e62bea5cf8b7adddaf3be6957d32b4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-14 16:41:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cde5a635fc052ffe53089a1e6c5d58a0ffdb1b98","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-14 18:00:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e6e9aa81612d4da408a8e8bb2143ac769a4056b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 18:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcddd5ab8288239ffe1dbe945195fe11d1fcbe79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 18:30:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"62baab716b57615e1d5ca136af940dc4945fd928","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 18:30:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d518308dea314ba782cf4565412633d5 : SUCCESS in 6m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b76a7fe1f58f4192b703e8addf49b97a : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66b2f41e5b4c4ff698b4e8ef2fea9d0b : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2df39cf6f796476091d4e800f7fb738f : SUCCESS in 14m 23s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5973cad5d8554cdc82fd11defefe7e4b : SUCCESS in 16m 57s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb891ab3e27c430ea2c59e3eca353db2 : SUCCESS in 15m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/486c78225f9b4eb4a998ef564eb3cbef : SUCCESS in 27m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e9e308aed184446aa751c70eda55e58 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2dc9e785a03e372afa0af8dad30313d958a47351","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 18:32:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf91df757ce4b6f9b99e21f4e4527af : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e21746f37588caffe9c0cb4c4a02e351f65a348f","revisions":{"e21746f37588caffe9c0cb4c4a02e351f65a348f":{"kind":"REWORK","_number":1,"created":"2020-07-14 11:28:55.000000000","uploader":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"ref":"refs/changes/42/740942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/42/740942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/740942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/740942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/42/740942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/42/740942/1"}}},"commit":{"parents":[{"commit":"80cfe119bbd750c09410cd3469ea6fb38b738eeb","subject":"Change to relative include","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/80cfe119bbd750c09410cd3469ea6fb38b738eeb"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 11:27:21.000000000","tz":600},"committer":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 11:28:52.000000000","tz":600},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I18e2f3376fefea600a1ad5a931ce52310b44f51f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e21746f37588caffe9c0cb4c4a02e351f65a348f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e21746f37588caffe9c0cb4c4a02e351f65a348f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
