)]}'
{"id":"openstack%2Fpuppet-octavia~748400","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fussuri~I53fc5ee6bc18d19d9108f73c8af8974eaff0bd43","project":"openstack/puppet-octavia","branch":"stable/ussuri","topic":"centos8-beaker-stable/ussuri","hashtags":[],"change_id":"I53fc5ee6bc18d19d9108f73c8af8974eaff0bd43","subject":"Use common class to manage apache service in beaker job","status":"MERGED","created":"2020-08-27 09:58:05.000000000","updated":"2020-08-27 14:57:11.000000000","submitted":"2020-08-27 14:57:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748400-1598540231851-8a9bf4a0","meta_rev_id":"133579c70506237377b58ab02f165cd39bba44bc","_number":748400,"virtual_id_number":748400,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-27 14:57:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-27 13:48:38.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-27 14:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-27 14:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 13:48:38.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-27 14:21:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-27 14:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c1773a81e6dbfb62bb6bf6b6e4e2f87e86fb7f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 09:58:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dbd9a24b35044411a8ca8d1114e6250fe3c2d508","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-27 10:10:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c41229e98db4532b856b2c1f5eff384885cea5eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 10:41:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eed1a1e5a21647aeb2ed350fc33d0647 : SUCCESS in 6m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/100ddf3f3d5340d080b9da0b75ac4c4a : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff75041cb86b4d75b947696edd2cf282 : SUCCESS in 4m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8ffc3d60caea418e9d8931a63e831e44 : SUCCESS in 19m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7052e768150746b4998300c155f51b3f : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcdf008ab44447cb88b2e397bc12486e : FAILURE in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9eb90b74753233efee7682d985c5271cbb191884","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-27 13:48:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"994715ea7986523ee5fcab5a992e086f4ad2a1cb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-27 14:21:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"287bcd390e4db2b3f15241c5839a60f40650162f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:22:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3a51b5fe501b28649a9ba23f6f26c9248d71c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:57:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71cbdf74141d49228cd2b80c7c591821 : SUCCESS in 7m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c322672a68248a584abca67c6c44c99 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/929167269e1d48389209308682a1aa8e : SUCCESS in 5m 38s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b0de25296c5a4d4480c960e425ee8986 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"41c80318cfbe1500a70650ed300a14d27be2e470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 14:57:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75c557562ef46abc6eba808cf7903aa14ff06c26","revisions":{"ad01b9ceb92fd8616f4da721eb2f589a319c9564":{"kind":"REWORK","_number":1,"created":"2020-08-27 09:58:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/748400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/00/748400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/1"}}},"commit":{"parents":[{"commit":"cfd750c5528f3ad19c580ba1f36fb7db8221c5d9","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cfd750c5528f3ad19c580ba1f36fb7db8221c5d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 03:05:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 09:58:05.000000000","tz":0},"subject":"Use common class to manage apache service in beaker job","message":"Use common class to manage apache service in beaker job\n\nDepends-on: https://review.opendev.org/#/c/745246/\nChange-Id: I53fc5ee6bc18d19d9108f73c8af8974eaff0bd43\n(cherry picked from commit 41c181975cba2ea963f9f213be4366b96d9a241f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ad01b9ceb92fd8616f4da721eb2f589a319c9564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ad01b9ceb92fd8616f4da721eb2f589a319c9564"}]},"branch":"refs/heads/stable/ussuri"},"75c557562ef46abc6eba808cf7903aa14ff06c26":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-27 10:10:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/748400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/00/748400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/00/748400/2"}}},"commit":{"parents":[{"commit":"cfd750c5528f3ad19c580ba1f36fb7db8221c5d9","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cfd750c5528f3ad19c580ba1f36fb7db8221c5d9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 03:05:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-27 10:10:10.000000000","tz":0},"subject":"Use common class to manage apache service in beaker job","message":"Use common class to manage apache service in beaker job\n\nDepends-on: https://review.opendev.org/#/c/747935/\nChange-Id: I53fc5ee6bc18d19d9108f73c8af8974eaff0bd43\n(cherry picked from commit 41c181975cba2ea963f9f213be4366b96d9a241f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/75c557562ef46abc6eba808cf7903aa14ff06c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/75c557562ef46abc6eba808cf7903aa14ff06c26"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
