)]}'
{"id":"openstack%2Fpuppet-octavia~748993","triplet_id":"openstack%2Fpuppet-octavia~master~I5eb1051e67a56941b9684728aa4f4340ddb35aab","project":"openstack/puppet-octavia","branch":"master","topic":"validate-db_conn","hashtags":[],"change_id":"I5eb1051e67a56941b9684728aa4f4340ddb35aab","subject":"Do not validate database_connection format","status":"MERGED","created":"2020-08-31 08:53:30.000000000","updated":"2020-08-31 14:03:39.000000000","submitted":"2020-08-31 14:03:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748993-1598882619075-f4e8b138","meta_rev_id":"d4221740fc219035aa2caa995aabbef10c050e70","_number":748993,"virtual_id_number":748993,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 14:03:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:27:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-31 08:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:27:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 08:55:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 13:27:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 14:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61841ec279c1446ea432ee31126a7ed5c5256247","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 08:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c700e47f4d0cc624bdf7ababf08fb105fec236f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 08:54:06.000000000","message":"Topic set to validate-db_conn","accounts_in_message":[],"_revision_number":1},{"id":"52891aa9ca7e01b136f1996afdf54b88bfa4daad","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 08:55:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a3f4b43d4025220db1dc28a30983f84e726874b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 11:17:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df7068e796934f599cffe6a379655176 : SUCCESS in 3m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afd41dcb409b47eb910bf4d3a2aa6762 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6073f336bad44f46afcf46da9fc81d31 : SUCCESS in 4m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9d06b6a6052f473dbe8ee1b7414f4e05 : SUCCESS in 13m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e03e9aa1860b4b72aee6d22e6e86d06f : SUCCESS in 13m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/273d0a9329e8408b8bdac1bfe2d51025 : SUCCESS in 8m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72a071f57a8b4a66ab9bfe1a2cfd2d7e : SUCCESS in 7m 58s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b253446ee8714532ad23bfd30e5ad8d5 : SUCCESS in 25m 41s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6b721773d0c40e5b9e8616711eda0c8 : SUCCESS in 13m 12s (non-voting)\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/99496bde25014294a512552406a3bca6 : SUCCESS in 2h 17m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3a7746dc18a7ef822fa1e10d3be2eb9bc455d01","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:27:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c99f3085aee9d5f6daf919d10807207db45a2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:27:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26559c6a72b7ca09ebddcdfd707ef623cc9fbff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 14:03:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49d94ad2d0f447cabf75bcad5f2aca76 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4297de03b9994bf0a8df74d1e4e8011b : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd6d8d3c1a3344108ac9a079865422cf : SUCCESS in 3m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e4e7cf7c12f0470db016cc5421f97ddf : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/43cfeef9de8947b0abfdd8c712ee9a2a : SUCCESS in 11m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34a8438ba02443f2b1b652af65e76ea9 : SUCCESS in 9m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a3eb38fb95e147df8a58ba1aff354df4 : SUCCESS in 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4221740fc219035aa2caa995aabbef10c050e70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 14:03:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4491d687b2e4310292bde62a1ce176a01d4d55df","revisions":{"4491d687b2e4310292bde62a1ce176a01d4d55df":{"kind":"REWORK","_number":1,"created":"2020-08-31 08:53:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/93/748993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/93/748993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/748993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/748993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/93/748993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/93/748993/1"}}},"commit":{"parents":[{"commit":"0c54350b8e497361ca7f48cbbed61691194b9ba9","subject":"Merge \"Add mysql_enable_ndb option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0c54350b8e497361ca7f48cbbed61691194b9ba9"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 08:53:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 08:53:21.000000000","tz":540},"subject":"Do not validate database_connection format","message":"Do not validate database_connection format\n\nCurrently we validate database_connection in 2 layers, each puppet\nmodules and puppet-oslo, however this makes it difficult to maintain\nvalidation pattern because we always need to fix both.\nThis patch removes the validation from each puppet modules so that\nwe need to maitain only one place, puppet-oslo to update validation\nlogic.\n\nChange-Id: I5eb1051e67a56941b9684728aa4f4340ddb35aab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4491d687b2e4310292bde62a1ce176a01d4d55df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4491d687b2e4310292bde62a1ce176a01d4d55df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
