)]}'
{"id":"openstack%2Fpuppet-octavia~760763","triplet_id":"openstack%2Fpuppet-octavia~master~I42c519f74d6b53570a8aaf9f4423d0a41ff4f0e4","project":"openstack/puppet-octavia","branch":"master","topic":"bug/1902158","hashtags":[],"change_id":"I42c519f74d6b53570a8aaf9f4423d0a41ff4f0e4","subject":"Deprecate allow_insecure_clients option","status":"MERGED","created":"2020-11-02 06:43:09.000000000","updated":"2020-11-03 09:56:26.000000000","submitted":"2020-11-03 09:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"760763-1604397242692-8dccae41","meta_rev_id":"c351229df723dce59f17f6f053f17ff8f4752bb1","_number":760763,"virtual_id_number":760763,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-03 02:02:00.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-03 09:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-11-03 06:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-03 09:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-11-03 09:15:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-03 06:34:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-11-03 09:15:36.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-11-03 09:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a51da5df63132b4de9bcb08d59d2b018ffcbd2d6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-02 06:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd32e9f9137fc97e02108b5077b5ee734c5ec8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 07:19:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/653dc049b4044be1a211e62794f65196 : FAILURE in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b4be447fe074bb0ac3c97106818c6ca : SUCCESS in 2m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8823961d3cf04df8a46178cc0b9963bf : SUCCESS in 9m 24s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aefd70ee4d5424cbd8e5568345cc3c1 : SUCCESS in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/300e2490a9f348ac9a9512cb0a0ad403 : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d2410839044c446f9068513d70bffdc7 : FAILURE in 13m 30s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f6e88447f75415ca665e799778cc7d4 : SUCCESS in 12m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c15b626e6e66478bb2e4d9b322f49b57 : POST_FAILURE in 6m 28s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/46e01f6707814f28a38d6d84d103fe2c : FAILURE in 13m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70cb2fbb66c70df149e434c7df4c0f25143b0ee8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-02 07:29:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ceae9912b22cc2dca1228c6bcd6625c1a560005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 07:54:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a0aa0000bc14c7cb2de274bbadd2123 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49a49d69dff141f38bc5f8e2abe4be4e : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/156e4ba8ee1f4c87a3d4f66db53f5274 : SUCCESS in 12m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f264c4460f84aa7a0cf3848c9d167c3 : SUCCESS in 10m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5180514a036849cc846d13d01c27acf7 : SUCCESS in 9m 00s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/ee0b2a3366e446c7b716839babafd12c : FAILURE in 12m 57s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ba649024e104ce4bbeb84b409887e1d : SUCCESS in 12m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2818aa7a9fd7429c911f0e70ab354ae7 : SUCCESS in 3m 49s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/227ebff1b0894970824ff2fccbe06c78 : FAILURE in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0757011c7aec7b535d371c5187420140baa2cd8","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-11-03 02:02:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9060cf99ef3b63ee5b2cd1f946c394f8106aea57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 02:55:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0be6b3d4bc1f4cd49069e62f26d1e62d : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4cdd9e98d734accb56228b958b95cf2 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/17749adbf8474f9b904e989adf9169f7 : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8989b57622cf439e97a1046b7e537862 : SUCCESS in 14m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac07e9eb7dd40f39afe8be6c066638d : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0e5aff82557e4187946c24c87497e488 : SUCCESS in 29m 59s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfb6e856127a4fac9e0a3e079016aeea : SUCCESS in 17m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21ec2103bc064406b9a512d9de3388d2 : SUCCESS in 3m 08s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fdcae2e6fe3a44e88c52bc78a54cd5ec : FAILURE in 14m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4447a4db124a339f752f860264ffddac17747a4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-03 06:34:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be3ce84fb4d2cf12f83b2193ddf7a7e26cfd2f37","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-03 09:15:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"29b1b2bac375dd7112db01bd82fd386fa3dbd390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:15:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d824f4e06a3eb91f1da9fb6c6fc26d93a2aea0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:54:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bd54c16e28731385a864799bac932d012c0735d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:54:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfe9193e84544701a2629c6933355f54 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/468015e03ccd46b3901186909c74c3e4 : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6167ac41d3894eb986a06eac096a2f73 : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e183eed5981641989f409f0383c3365a : SUCCESS in 8m 04s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/494f2edf3e664698ab83ac4f507b9468 : SUCCESS in 24m 02s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d18c79d3d95d4338afb12cbb454e2d0f : SUCCESS in 12m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5feae0eabda48558c71caaa4d4b835b : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"22aedf6cfde3eecc352fe3d84d6bfba560223b22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 09:56:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82f6abdf8b33472cbbeb854d683e64a3 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a8940c87e782ad437ff8e20d8f5dbfec8ec6109","revisions":{"978eb155b12d22e75a0f717bca960db11cf6ebee":{"kind":"REWORK","_number":1,"created":"2020-11-02 06:43:09.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/63/760763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/63/760763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/1"}}},"commit":{"parents":[{"commit":"87e27164954196944f6a4a0f42f71427afeebbe9","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/87e27164954196944f6a4a0f42f71427afeebbe9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:04.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:04.000000000","tz":480},"subject":"Deprecate allow_insecure_clients option","message":"Deprecate allow_insecure_clients option\n\nThe allow_insecure_clients has been deprecated[1].\n\n[1]https://review.opendev.org/#/c/417629/\n\nChange-Id: I42c519f74d6b53570a8aaf9f4423d0a41ff4f0e4\nCloses-Bug: #1902158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/978eb155b12d22e75a0f717bca960db11cf6ebee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/978eb155b12d22e75a0f717bca960db11cf6ebee"}]},"branch":"refs/heads/master"},"8a8940c87e782ad437ff8e20d8f5dbfec8ec6109":{"kind":"REWORK","_number":2,"created":"2020-11-02 07:29:20.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/63/760763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/63/760763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/63/760763/2"}}},"commit":{"parents":[{"commit":"87e27164954196944f6a4a0f42f71427afeebbe9","subject":"Fix outdated files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/87e27164954196944f6a4a0f42f71427afeebbe9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 06:42:04.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-11-02 07:29:01.000000000","tz":480},"subject":"Deprecate allow_insecure_clients option","message":"Deprecate allow_insecure_clients option\n\nThe allow_insecure_clients has been deprecated[1].\n\n[1]https://review.opendev.org/#/c/417629/\n\nChange-Id: I42c519f74d6b53570a8aaf9f4423d0a41ff4f0e4\nCloses-Bug: #1902158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8a8940c87e782ad437ff8e20d8f5dbfec8ec6109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8a8940c87e782ad437ff8e20d8f5dbfec8ec6109"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
