)]}'
{"id":"openstack%2Fpuppet-octavia~772180","triplet_id":"openstack%2Fpuppet-octavia~master~I63a585b34eaeb5350dbbd72936b4949ebd96e0f0","project":"openstack/puppet-octavia","branch":"master","topic":"quota-param-names","hashtags":[],"change_id":"I63a585b34eaeb5350dbbd72936b4949ebd96e0f0","subject":"Use consistent parameter names for quota options","status":"MERGED","created":"2021-01-24 02:21:57.000000000","updated":"2021-01-25 18:30:42.000000000","submitted":"2021-01-25 18:28:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772180-quota-param-names","meta_rev_id":"9b086395700a7fbecd3f43bf18005b095850fe7d","_number":772180,"virtual_id_number":772180,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 18:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-01-24 22:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-01-25 18:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-01-25 18:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-24 12:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-24 22:13:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-01-25 18:01:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"95a9d0b71eac235938b56e0f275f3d7133d48dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 02:21:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"835f2c71221f0c5fcc7332ca15c49cb4677d294f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 02:23:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a141d959ee2fa390b3dba4b5cdaabe52537e47c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 02:24:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4817965ae51229692c1b8fec1779429d8bee39c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 04:16:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5916893af2edbf7abe60501e66588b761f60761a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 07:06:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"407ba8c07fc846ec8f04d16367f9762f0a028a1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-24 08:47:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a33b36f4fce2bf3bacdae4305a6b14a5a38ead83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 12:03:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca148d058a4b4110b5f3d87417d4372b : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b9c963bf71246e1a969d21d0eb5ea7e : SUCCESS in 2m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3cc6506786684ff0a90e54689a69f2ee : SUCCESS in 8m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca96615ca82a4137a493b757f93df9c0 : SUCCESS in 6m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aecffb202d349cea90292283b0a1cf5 : SUCCESS in 7m 19s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b5e7d275d7244e4db425d37d03cc0004 : SUCCESS in 22m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72427a618a60435ebfb867a4a58742c7 : SUCCESS in 12m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4b654a26bc34f85965547b7e57828ef : SUCCESS in 3m 19s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/72597d0534894f23acfd0d4f398f48dd : TIMED_OUT in 3h 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6d5b883dac71f2e26e2a8e62315273ecf78c2648","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-01-24 22:13:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c176b9f174c2ea2c2256c09fa63fee8c64de3a6f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-25 18:01:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e8ae2b5952509880cf66aec5cc593e422e63ccc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:02:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"33e74b91629d990bf9b1ea6dae12e84dcc6f7472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:28:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8303040eef54447189307fe2b22b4c89 : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe4fd5fc32d5451fb9622fd45f3da3b5 : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d976d834b66e43cc8150949ead1c186f : SUCCESS in 8m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ba62f97da10457f9330c9459a5fcfcb : SUCCESS in 7m 01s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9a3564ddbc3e4d26bf6e64f5482df772 : SUCCESS in 22m 31s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/288d4b6b573841229dc6288b53d689ff : SUCCESS in 11m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886f6f60997542f4ac907e073965cb32 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"d5dd8066c999ae08326254b171ca1f4471d92ce4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:28:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"9b086395700a7fbecd3f43bf18005b095850fe7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 18:30:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaaa6b097f9e4f13b9b5e2686424dda2 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7d5b8aafc8fdce1f3d8f44e6fa00479811260c50","revisions":{"b68730f031ef8b034dc1e68a9afd2fb800194324":{"kind":"REWORK","_number":1,"created":"2021-01-24 02:21:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/1"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b68730f031ef8b034dc1e68a9afd2fb800194324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b68730f031ef8b034dc1e68a9afd2fb800194324"}]},"branch":"refs/heads/master"},"ff292e280ed5f9a8cad078d776298d69f556a51a":{"kind":"REWORK","_number":2,"created":"2021-01-24 02:23:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/2"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:23:27.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff292e280ed5f9a8cad078d776298d69f556a51a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff292e280ed5f9a8cad078d776298d69f556a51a"}]},"branch":"refs/heads/master"},"aa99b5cba225022026799205af3b0595d5458005":{"kind":"REWORK","_number":3,"created":"2021-01-24 02:24:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/3"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:24:40.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/aa99b5cba225022026799205af3b0595d5458005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/aa99b5cba225022026799205af3b0595d5458005"}]},"branch":"refs/heads/master"},"e655784aebaea9b03978698bcae184e9df05848f":{"kind":"REWORK","_number":4,"created":"2021-01-24 04:16:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/4"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 04:16:09.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e655784aebaea9b03978698bcae184e9df05848f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e655784aebaea9b03978698bcae184e9df05848f"}]},"branch":"refs/heads/master"},"a0a2c575ee4d034de0bd40f59c785b8f7c4b18fe":{"kind":"REWORK","_number":5,"created":"2021-01-24 07:06:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/5"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 07:06:48.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a0a2c575ee4d034de0bd40f59c785b8f7c4b18fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a0a2c575ee4d034de0bd40f59c785b8f7c4b18fe"}]},"branch":"refs/heads/master"},"7d5b8aafc8fdce1f3d8f44e6fa00479811260c50":{"kind":"REWORK","_number":6,"created":"2021-01-24 08:47:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/772180/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/80/772180/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/80/772180/6"}}},"commit":{"parents":[{"commit":"46cffbd1e9229f587f3b6c703f4964aec4bab452","subject":"Deprecate parameters for the Spares Pool feature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/46cffbd1e9229f587f3b6c703f4964aec4bab452"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 02:20:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-24 08:47:35.000000000","tz":540},"subject":"Use consistent parameter names for quota options","message":"Use consistent parameter names for quota options\n\nThis change renames parameters of octavia::quota options, so that we\nuse consistent parameter names with the definitions in Octavia.\n\nChange-Id: I63a585b34eaeb5350dbbd72936b4949ebd96e0f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7d5b8aafc8fdce1f3d8f44e6fa00479811260c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7d5b8aafc8fdce1f3d8f44e6fa00479811260c50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
