)]}'
{"id":"openstack%2Fpuppet-octavia~773279","triplet_id":"openstack%2Fpuppet-octavia~master~I7359d476e7006d9c02cd2722a716f515f1cf9b55","project":"openstack/puppet-octavia","branch":"master","topic":"healthcheck","hashtags":[],"change_id":"I7359d476e7006d9c02cd2722a716f515f1cf9b55","subject":"Add support for healthcheck middleware options","status":"MERGED","created":"2021-02-01 03:25:49.000000000","updated":"2021-03-26 04:58:46.000000000","submitted":"2021-03-26 04:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773279-healthcheck","meta_rev_id":"0729285022d43fe10398afade69854a946204393","_number":773279,"virtual_id_number":773279,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 04:57:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-03-26 04:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-03-09 15:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-03-26 04:25:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 06:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 17:56:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-01 13:14:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-26 04:25:37.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"017088a5e3f797ab55ba2ff91f1ea6b4630405e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-01 03:25:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a0f3e1d03aea4804f561ef6986a231c7a8d24d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 06:45:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/528bcdd4fa6c41ef8d4c00adf4e6fb4a : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/003015c41ac14839bd58dfdcdba4c0e4 : SUCCESS in 2m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/df1452394c9348d2b29b193cac3721bd : FAILURE in 7m 05s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9570ea7efe94da793442aa6d7dbb7fd : FAILURE in 7m 03s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f01cb9d0eb741e9abdb25a90cde179f : FAILURE in 5m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d1caf2fdcae4c8d8101305f2e7d6fdd : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8f78f9b9577c4629a1b9f92316bdbf43 : SUCCESS in 23m 22s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a90229db81f40b280318c406443071a : SUCCESS in 11m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e777f3e367734a74afbd32a6fb802adc : SUCCESS in 3m 10s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a0d4a6614055467b9790ba7443d90e64 : TIMED_OUT in 3h 14m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee002107eaea80f083e42ce8166b6e0350cc1a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-01 07:30:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fdfe192610975d090c8f7903eac6db7473c2efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 10:44:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/feca16a817314f2680c9b18d2c938c1d : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef8a81848f0847338d017af6af1954c6 : SUCCESS in 3m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/09890a88bdfc412f8365b03ca93ad31d : SUCCESS in 8m 06s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/752aa80b08f74ed4aefb628a38a105e2 : SUCCESS in 10m 54s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbc0858e79584019a6ac063f3e5270c5 : SUCCESS in 6m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e82df43ea7bd460d9fe3c19e089f9889 : SUCCESS in 7m 03s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a4870cf0f566483bb663035879d29cd6 : SUCCESS in 21m 40s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49975264841e49d997f8c3c80f8b0708 : SUCCESS in 10m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d59dd3422904d998f891704bd2a5280 : SUCCESS in 3m 22s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e7f46cbc79d542a18bd9c66b19372236 : TIMED_OUT in 3h 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c0b20705cbb838b3b374430d27191dc33938ee7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:56:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"505eaf27a722098d6dac28d5c18ce6de013f8dbf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-01 13:14:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cebcc8e6688b4f077e98b0954f2f4ff16cae5ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 13:15:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f352f3c69aac72a75f538d22a59e5738d0a50a5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 13:42:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc765e7a51564b198e500b5f8c51b106 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98f5b9ef845d410b88c66225c6c698c5 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b9e1191442874ce9940a7f4c29fb110e : FAILURE in 13m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c497d4b890af4323a7d6915582fcf322 : FAILURE in 10m 09s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/2670e8c2d08a4b9fa937ed1834bad4b4 : SUCCESS in 25m 02s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b76d31cdce874d49b747e6377e0a92e0 : SUCCESS in 15m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6401612221f04dc88dd71ffd0bec97cb : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"89674647dc8e9d37706b9e18694d5beb3f4d0abf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-04 04:02:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d7d56bd832b4468faea1dd76ca8ce7343de5b424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 07:17:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7e222a999ee4afd99b13d5f8b775258 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb285d3f64464e35853a29d26fb68692 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0896bb05612f44c5853cbc982795daa6 : FAILURE in 12m 01s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/60eed12eea3a41fb8e6041af3339a38f : FAILURE in 9m 44s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf54dd01154941419b1eb389c4e54660 : FAILURE in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5069be6d45e249159e26e50f61055f09 : FAILURE in 7m 56s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/551bb93b492f47ea95b5fc34906db6c9 : SUCCESS in 22m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/57de9356549a44a3a0656dc0b14b250a : SUCCESS in 22m 51s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab81dfc2a62d4889b54079bbb446eaf9 : SUCCESS in 13m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b4ac7fff729421091eba77fb38c10c9 : SUCCESS in 4m 06s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eec1ad04f5344d9d922ccbd3239af103 : TIMED_OUT in 3h 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06042d8dffece625db2f835675ab0b55fe4da615","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-05 07:45:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d6622ab1b88a244558bef5aa52a6d0705a0d1d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:56:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff38d12b6ba14a54ac8637ff11808548 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebce6ee80afa4d17b0d045a76fe92712 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/288d6d3adb5c4ad1885fa31d37a211aa : FAILURE in 8m 22s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb148ff9b7f74d44bbac382e693322a4 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8216dcf37ac24f42a4f5b6c9e8020f06 : FAILURE in 6m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf98a9693d6e44719db4e9780cf5ddf4 : FAILURE in 7m 11s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/cf15058a459a447fbdd9a1d5e7a8144c : SUCCESS in 21m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7c07ee13a164fec9407846e633f66e9 : SUCCESS in 19m 16s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11503568aaea48548626b02b0b274a04 : SUCCESS in 14m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a07fa0c225e64fe2ac9bc227baf0a5b2 : SUCCESS in 3m 21s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3c335e1cc8eb4070883db8ad59a54823 : FAILURE in 3h 07m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57dfeb5d792e85e5ae284473d25410df20802505","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-05 13:24:11.000000000","message":"Patch Set 2:\n\nWe need https://review.opendev.org/c/openstack/puppet-octavia/+/778843 to fix the broken unit tests.","accounts_in_message":[],"_revision_number":2},{"id":"fa07dcbe45279ffdc6cd97dd047da23c1f52eab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-09 11:41:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"90d372a841d08516ccd62cd99e82da5727d07a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 14:57:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f23d9c241ffb4f6cb23031380f590df6 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5de4f60b90b04fb6be4b8098d8ae91b2 : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e7afe8e8ee15405d95a1d4ffc436d6e2 : SUCCESS in 10m 25s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a854cc47c5be41cbb8e8ea973b85f940 : SUCCESS in 9m 29s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d47029c75d449dc80db4cbd45e22e9f : SUCCESS in 8m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af01c1a24fc943e9aa755a9902366a59 : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4f823caf1fd245e78d47c6c0767e7e8f : SUCCESS in 47m 10s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/3428d3e24f8448c7b929c513253c1060 : SUCCESS in 24m 52s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d22edabdec44d1faef9e37f33c42863 : SUCCESS in 16m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f7c140a08f94db9860e6a6b0444a111 : SUCCESS in 5m 18s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5659de72a5c14bce9e1c8c246edf09ff : TIMED_OUT in 3h 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"658824eae366eccf0914080235319b3eaaf114b1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-09 15:11:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2cd73fc3e056fc4cea04e69e49cae44dcb49c97e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-26 04:25:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"41b86f4d22bfb3f2983c04816556c9dd99e97233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:26:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0e1eb6ede1bae809f54646c6bcee274c301d5e18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:57:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a6369ea289f416f907687f48cab426f : SUCCESS in 4m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdd13e6308184eb2b806e6620206b95f : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5f4fb48e9e142c3a055e6a54921e9a4 : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ac7f337c6244f8b9fe3990c5dccbfea : SUCCESS in 8m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/264e572d87424b19bb867eedf51810c0 : SUCCESS in 25m 49s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22f41fca0e5441cda2470e185885bd3f : SUCCESS in 14m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6a0982c7c248c19b4c7c53bd3aedab : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c0aa92d4b5f28960957aee9c05494d8c40c2d1fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:57:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0729285022d43fe10398afade69854a946204393","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 04:58:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239498a0b809426187ba88650ad452ba : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"66604db840f473b0aa0a4460e5b12bdb10c4c2d8","revisions":{"e38b29c4763a8a2df1568e87ba364dd26e6fe907":{"kind":"REWORK","_number":1,"created":"2021-02-01 03:25:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/773279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/773279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/1"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 03:25:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 03:25:31.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7359d476e7006d9c02cd2722a716f515f1cf9b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e38b29c4763a8a2df1568e87ba364dd26e6fe907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e38b29c4763a8a2df1568e87ba364dd26e6fe907"}]},"branch":"refs/heads/master"},"c1ace023231f6d53a01fbfe716c31c12b5aa8b75":{"kind":"REWORK","_number":2,"created":"2021-02-01 07:30:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/773279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/773279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/2"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 03:25:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 07:30:49.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7359d476e7006d9c02cd2722a716f515f1cf9b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c1ace023231f6d53a01fbfe716c31c12b5aa8b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c1ace023231f6d53a01fbfe716c31c12b5aa8b75"}]},"branch":"refs/heads/master"},"66604db840f473b0aa0a4460e5b12bdb10c4c2d8":{"kind":"REWORK","_number":3,"created":"2021-03-09 11:41:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/773279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/773279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/773279/3"}}},"commit":{"parents":[{"commit":"1217a8fc35b78e37908dacc02a4e8dc42628ac89","subject":"Add support for the api_settings/pagination_max_limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1217a8fc35b78e37908dacc02a4e8dc42628ac89"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-01 03:25:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-09 11:40:08.000000000","tz":540},"subject":"Add support for healthcheck middleware options","message":"Add support for healthcheck middleware options\n\nDepends-on: https://review.opendev.org/772202\nChange-Id: I7359d476e7006d9c02cd2722a716f515f1cf9b55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/66604db840f473b0aa0a4460e5b12bdb10c4c2d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/66604db840f473b0aa0a4460e5b12bdb10c4c2d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
