)]}'
{"id":"openstack%2Fpuppet-octavia~778836","triplet_id":"openstack%2Fpuppet-octavia~master~Id8c389c136de34fd9d6d9fd762a3bf96f5d02267","project":"openstack/puppet-octavia","branch":"master","topic":"octavia-pagination","hashtags":[],"change_id":"Id8c389c136de34fd9d6d9fd762a3bf96f5d02267","subject":"Add support for the api_settings/pagination_max_limit","status":"MERGED","created":"2021-03-05 03:33:24.000000000","updated":"2021-03-09 03:28:46.000000000","submitted":"2021-03-09 03:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"778836-octavia-pagination","meta_rev_id":"c4e298b403170eb5d69be4294ff697f2ce6f138d","_number":778836,"virtual_id_number":778836,"owner":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 03:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-03-05 13:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-05 18:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-03-05 18:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 04:46:17.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-05 10:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 18:05:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f0bcc1bc9ad27b0d27828e6b306aa8cfb578240f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2021-03-05 03:33:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2373693bd892f6963a6552e2e709115a40c7c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2021-03-05 03:35:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"930b613cadff15405bcd1585bf610e09329ea9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2021-03-05 04:40:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ffe282165d595d773cf5addec52d57b887c9cd5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-05 04:56:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"67cc68e263732c66b5f45b1525df808a73542dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2021-03-05 05:17:55.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3d8ae061dec52ec6e084163b395541b481fd8875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"date":"2021-03-05 07:40:14.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"39a1bc41a15fc56fe89cb52ce0216b5888315b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 10:55:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f88bf2bb2d6448f49f6111d81ef96629 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9f92fa7ced04a7286406dbda495b9c0 : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4e95778b6ed249d48dc04669bb2e1f7b : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/02ec9022942c4727ba7c91dd96225c49 : SUCCESS in 10m 12s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8aa7b93f24da4cc78083aff97c2cc5fc : SUCCESS in 9m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9080e82074d40b3832497e1a02bd515 : SUCCESS in 6m 46s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c8cef32612234ca7995da0ae9e11149d : SUCCESS in 18m 59s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/21c52555c23f4412b1afce401c3f5c25 : SUCCESS in 27m 17s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/928eb5159d14454ab7618d1c6b7f435b : SUCCESS in 16m 18s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5662492471134aba811ebb3631415224 : SUCCESS in 3m 38s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/06d83e1cf5a044e895c8552ac5335406 : TIMED_OUT in 3h 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e0bdf2ab5aacf330929ba64e2faf2d9e3cdb1224","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-05 13:25:27.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nLGTM. Thanks !","accounts_in_message":[],"_revision_number":5},{"id":"b2d2fc7c904d8f0b071d93d2e3c6b91fdb2f4762","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-05 18:05:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5a2bb38fefe16d95d82fe72ebb9d9f5f1baf950e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 02:57:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c1b164f0c7e3f1af53182d22287577fe569a6a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:27:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8de15fb7b064e3dab568a69d54d26a1 : SUCCESS in 3m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24352c01bd384d479e8edde53529ff74 : SUCCESS in 4m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/540827c7982349fdbc15f7555b6abf6f : SUCCESS in 10m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab466482c44049aba22a4c27bf8140a9 : SUCCESS in 8m 03s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9e9838ca3e0c4e8eacb45cd84666c114 : SUCCESS in 24m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/71fb827907ba48d3b913c432ea385106 : SUCCESS in 22m 00s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe12ae10c02941cb90bb4b5953241416 : SUCCESS in 13m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2aa65145f14b4cfaa79ecf02c9f91505 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":5},{"id":"be853278c8f03211d9d1726c01d439e36081ceb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c4e298b403170eb5d69be4294ff697f2ce6f138d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 03:28:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f4d152ffd7d429ea8df786849b472aa : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1217a8fc35b78e37908dacc02a4e8dc42628ac89","revisions":{"576e8aa3a86a9db9ca52cd09d569d6b6cc07ba18":{"kind":"REWORK","_number":1,"created":"2021-03-05 03:33:24.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/36/778836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/778836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/1"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-04 10:46:12.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-05 03:33:20.000000000","tz":600},"subject":"Add support for the keystone_api_settings/pagination_max_limit","message":"Add support for the keystone_api_settings/pagination_max_limit\n\nThis patch implements a new parameter named pagination_max_limit into\noctaiva::api_settings::pagination_max_limit, so that operators can\nconfigure that parameter by puppet_octaiva.\n\nChange-Id: Id8c389c136de34fd9d6d9fd762a3bf96f5d02267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/576e8aa3a86a9db9ca52cd09d569d6b6cc07ba18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/576e8aa3a86a9db9ca52cd09d569d6b6cc07ba18"}]},"branch":"refs/heads/master"},"f950d201f42cb282debfd24a93737ac01918d5aa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-05 03:35:43.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/36/778836/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/778836/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/2"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-04 10:46:12.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-05 03:35:11.000000000","tz":600},"subject":"Add support for the octavia_api_settings/pagination_max_limit","message":"Add support for the octavia_api_settings/pagination_max_limit\n\nThis patch implements a new parameter named pagination_max_limit into\noctaiva::api_settings::pagination_max_limit, so that operators can\nconfigure that parameter by puppet_octaiva.\n\nChange-Id: Id8c389c136de34fd9d6d9fd762a3bf96f5d02267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f950d201f42cb282debfd24a93737ac01918d5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f950d201f42cb282debfd24a93737ac01918d5aa"}]},"branch":"refs/heads/master"},"8f724c2ec3da92e6bae6ab26d023c81ba6c89d2a":{"kind":"REWORK","_number":3,"created":"2021-03-05 04:40:03.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/36/778836/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/778836/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/3"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-04 10:46:12.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-05 04:38:32.000000000","tz":600},"subject":"Add support for the octavia_api/pagination_max_limit","message":"Add support for the octavia_api/pagination_max_limit\n\nThis patch implements a new parameter named pagination_max_limit into\noctaiva::api::pagination_max_limit, so that operators can configure\nthat parameter by puppet-octaiva.\n\nChange-Id: Id8c389c136de34fd9d6d9fd762a3bf96f5d02267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8f724c2ec3da92e6bae6ab26d023c81ba6c89d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8f724c2ec3da92e6bae6ab26d023c81ba6c89d2a"}]},"branch":"refs/heads/master"},"7c4290e7ace387d260917ac6f1ab61797556b711":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-05 05:17:55.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/36/778836/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/778836/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/4"}}},"commit":{"parents":[{"commit":"07d702b0c95779d9e23346abef217656eb8d7340","subject":"Merge \"Use consistent parameter names for quota options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/07d702b0c95779d9e23346abef217656eb8d7340"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-04 10:46:12.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-05 05:13:25.000000000","tz":600},"subject":"Add support for the api_settings/pagination_max_limit","message":"Add support for the api_settings/pagination_max_limit\n\nThis patch implements a new parameter named pagination_max_limit into\noctaiva::api, so that operators can configure that parameter by puppet-octaiva.\n\nChange-Id: Id8c389c136de34fd9d6d9fd762a3bf96f5d02267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c4290e7ace387d260917ac6f1ab61797556b711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7c4290e7ace387d260917ac6f1ab61797556b711"}]},"branch":"refs/heads/master"},"1217a8fc35b78e37908dacc02a4e8dc42628ac89":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-05 07:40:14.000000000","uploader":{"_account_id":30075,"name":"Christopher Brown","email":"chrisbro@redhat.com","username":"chrisbro"},"ref":"refs/changes/36/778836/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/36/778836/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/36/778836/5"}}},"commit":{"parents":[{"commit":"be9157566e0025caf5cc51eb49aee24c8e623ce0","subject":"Do not expect database/idle_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/be9157566e0025caf5cc51eb49aee24c8e623ce0"}]}],"author":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-04 10:46:12.000000000","tz":600},"committer":{"name":"Christopher Brown","email":"chrisbro@redhat.com","date":"2021-03-05 07:40:14.000000000","tz":0},"subject":"Add support for the api_settings/pagination_max_limit","message":"Add support for the api_settings/pagination_max_limit\n\nThis patch implements a new parameter named pagination_max_limit into\noctaiva::api, so that operators can configure that parameter by puppet-octaiva.\n\nChange-Id: Id8c389c136de34fd9d6d9fd762a3bf96f5d02267\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1217a8fc35b78e37908dacc02a4e8dc42628ac89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1217a8fc35b78e37908dacc02a4e8dc42628ac89"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
