)]}'
{"id":"openstack%2Fpuppet-octavia~783162","triplet_id":"openstack%2Fpuppet-octavia~master~I5805c762d0dd6a21a2b845858fbb2653e843c60b","project":"openstack/puppet-octavia","branch":"master","topic":"prep-release-wallaby-rc1","hashtags":[],"change_id":"I5805c762d0dd6a21a2b845858fbb2653e843c60b","subject":"Prepare Wallaby RC1","status":"MERGED","created":"2021-03-26 03:26:25.000000000","updated":"2021-03-26 09:03:26.000000000","submitted":"2021-03-26 09:03:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783162-prep-release-wallaby-rc1","meta_rev_id":"b33f0c2b4fea42ca1e32dc4635294862a19f8a11","_number":783162,"virtual_id_number":783162,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 09:03:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-03-26 05:04:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-26 08:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-03-26 08:29:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 03:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 04:15:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-26 08:29:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"df917278e79d59a153da755db4dce5075cbced69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-03-26 03:26:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df4585ed487d93b0bdf3cde093a53d6801f5703f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 03:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/86e00fb53933437fab0592609f6fc255 : SUCCESS in 3m 15s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5125969d731b4ecb80d9fe0cf97119bb : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a560255bef80f9a17ebb35933a7fa266272b41d9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 04:15:43.000000000","message":"Patch Set 1: Code-Review-1\n\nWould be nice if we can merge some patches waiting for second +2...\n\nhttps://review.opendev.org/c/openstack/puppet-octavia/+/780256\nhttps://review.opendev.org/c/openstack/puppet-octavia/+/780259\nhttps://review.opendev.org/c/openstack/puppet-octavia/+/773279","accounts_in_message":[],"_revision_number":1},{"id":"05e091ab8d97c9439212d169a985af6439962f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 05:03:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b603276ad89bfcfdb20d7924a50ede640c009b9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 05:04:16.000000000","message":"Patch Set 2: Code-Review+2\n\nRebased on master because now the said patches have been merged.","accounts_in_message":[],"_revision_number":2},{"id":"141d71df3919a0076e7c10a73fdffceb7f9c8284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 05:11:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b9afd03b9e6e4551bc4e4aa21ba07b18 : SUCCESS in 2m 54s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0af7578ad014fa3a7d9c6a797ca250f : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"5e6057330ec457605b5eaf7f7d2ec588e7e4bdf7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-26 08:29:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"79068b5cb509ba512b5d6433f2788f2820d4550a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 08:46:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"43009c883c8bb83720bf3bdcdd4e026083948549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 09:03:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/97d3a1d0283c42788f269fb5b1fbe77c : SUCCESS in 2m 47s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b240559413334ac58a1c73712ec8d0c4 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b33f0c2b4fea42ca1e32dc4635294862a19f8a11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 09:03:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"793451b46898ebe3e8cc6b42c94813604e7b47ec","revisions":{"b833af5ff352627b49eb35863f6127c1c43f1608":{"kind":"REWORK","_number":1,"created":"2021-03-26 03:26:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/62/783162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/62/783162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/1"}}},"commit":{"parents":[{"commit":"613def36aa39aba8bc24290a15223bf495fbd009","subject":"Merge \"Add support for oslo_policy/enforce_new_defaults\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/613def36aa39aba8bc24290a15223bf495fbd009"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-26 03:24:16.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-26 03:24:16.000000000","tz":480},"subject":"Prepare Wallaby RC1","message":"Prepare Wallaby RC1\n\nUpdate the version metadata for Wallaby milestone RC1\n\nChange-Id: I5805c762d0dd6a21a2b845858fbb2653e843c60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b833af5ff352627b49eb35863f6127c1c43f1608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b833af5ff352627b49eb35863f6127c1c43f1608"}]},"branch":"refs/heads/master"},"793451b46898ebe3e8cc6b42c94813604e7b47ec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-26 05:03:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/783162/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/62/783162/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/62/783162/2"}}},"commit":{"parents":[{"commit":"11796ca87c963ca3921c5930db1a5b641fff33b6","subject":"Merge \"Add support for [controller_worker] image_driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/11796ca87c963ca3921c5930db1a5b641fff33b6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-03-26 03:24:16.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 05:03:29.000000000","tz":0},"subject":"Prepare Wallaby RC1","message":"Prepare Wallaby RC1\n\nUpdate the version metadata for Wallaby milestone RC1\n\nChange-Id: I5805c762d0dd6a21a2b845858fbb2653e843c60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/793451b46898ebe3e8cc6b42c94813604e7b47ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/793451b46898ebe3e8cc6b42c94813604e7b47ec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
