)]}'
{"id":"openstack%2Fpuppet-octavia~786223","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~Ib4077b4c89234b024d027a4195c40b7b2d71acaa","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"dnm-stable/wallaby","hashtags":[],"change_id":"Ib4077b4c89234b024d027a4195c40b7b2d71acaa","subject":"DNM: Test litmus job","status":"ABANDONED","created":"2021-04-14 14:47:27.000000000","updated":"2022-01-13 05:03:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":786253,"cherry_pick_of_patch_set":1,"meta_rev_id":"4df609d44498db0153565e9150185ec54aa70c61","_number":786223,"virtual_id_number":786223,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-14 18:04:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 18:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e00293913e23671aabfca54b7bbe09ae3a33d7cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:47:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1da99ca824685502bbd895639ccdda5575d66e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:04:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7b17ae9926462489a8cd1aa6059ae7 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7050090d9cdf421282e2743a83b22d4a : SUCCESS in 4m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/24f7852255ad4a96ac8cbbb7d84fb965 : SUCCESS in 10m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dcc0f116b414699a4442295d6940405 : SUCCESS in 7m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139c83fe34ad4cea9f1fb9b520af7b26 : SUCCESS in 10m 24s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a6d73cfe39b45e2bde04775986ab96d : FAILURE in 28m 06s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb2295d9541e495286677f8881853e07 : SUCCESS in 14m 11s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7d3be3d52b754f45ab204fabc9cb1231 : TIMED_OUT in 3h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4df609d44498db0153565e9150185ec54aa70c61","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-13 05:03:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6af4b323ffbdfe4291b46c51eb2f51198722bbe","revisions":{"a6af4b323ffbdfe4291b46c51eb2f51198722bbe":{"kind":"REWORK","_number":1,"created":"2021-04-14 14:47:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/786223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/23/786223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/786223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/786223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/23/786223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/23/786223/1"}}},"commit":{"parents":[{"commit":"55ce60902b04ebc0e219ae108f13762896afd307","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/55ce60902b04ebc0e219ae108f13762896afd307"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-14 14:41:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-14 14:47:27.000000000","tz":0},"subject":"DNM: Test litmus job","message":"DNM: Test litmus job\n\nChange-Id: Ib4077b4c89234b024d027a4195c40b7b2d71acaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a6af4b323ffbdfe4291b46c51eb2f51198722bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a6af4b323ffbdfe4291b46c51eb2f51198722bbe"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
