)]}'
{"id":"openstack%2Fpuppet-octavia~786224","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fvictoria~Ib4077b4c89234b024d027a4195c40b7b2d71acaa","project":"openstack/puppet-octavia","branch":"stable/victoria","topic":"dnm-stable/wallaby-stable/victoria","hashtags":[],"change_id":"Ib4077b4c89234b024d027a4195c40b7b2d71acaa","subject":"DNM: Test litmus job","status":"ABANDONED","created":"2021-04-14 14:47:47.000000000","updated":"2021-04-18 04:46:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":786223,"cherry_pick_of_patch_set":1,"meta_rev_id":"980c0ed08f689c57abfc2fa08504b382e45eadc4","_number":786224,"virtual_id_number":786224,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-14 16:58:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 16:58:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d0b11f6031069d86b78e2a0ca217bd89dd3cdb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:47:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"e8a07421b4291f15231e633b6545e059b4587c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:58:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae231b16a5b848b182de94b2f97f1569 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff81621c5d8a4fb7af8aad998836ac3f : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/60570bc9a0a342fdb0f0fc6d6c283e45 : SUCCESS in 13m 14s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfaa946427c64110b32cee76aad94de5 : SUCCESS in 9m 48s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3aeaadddf2ae44bb8dcb461e585d7833 : SUCCESS in 8m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17e6c4f263464b3c8d66cccd7fbdcbbb : SUCCESS in 11m 04s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/49a5d0150db34fc58d2d28b7cc1becb8 : SUCCESS in 23m 42s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc77973be91348e583cf00d36d888e12 : SUCCESS in 24m 28s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83d0d383d9e543e58d727bff2d7a4a3d : SUCCESS in 14m 22s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/91543a5ed51c48c39a3a206f969ed518 : FAILURE in 2h 06m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"980c0ed08f689c57abfc2fa08504b382e45eadc4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-18 04:46:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31408dbda6d5c68d276af3d485287eded3cb0c57","revisions":{"31408dbda6d5c68d276af3d485287eded3cb0c57":{"kind":"REWORK","_number":1,"created":"2021-04-14 14:47:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/786224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/24/786224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/786224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/786224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/24/786224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/24/786224/1"}}},"commit":{"parents":[{"commit":"0a6a63692cd0bceae99248aee3f591752dba269c","subject":"Do not expect database/idle_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0a6a63692cd0bceae99248aee3f591752dba269c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-14 14:41:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-14 14:47:47.000000000","tz":0},"subject":"DNM: Test litmus job","message":"DNM: Test litmus job\n\nChange-Id: Ib4077b4c89234b024d027a4195c40b7b2d71acaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/31408dbda6d5c68d276af3d485287eded3cb0c57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/31408dbda6d5c68d276af3d485287eded3cb0c57"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
