)]}'
{"id":"openstack%2Fpuppet-octavia~789526","triplet_id":"openstack%2Fpuppet-octavia~master~I07b12acbaba8eb8affc603c2679297495230d37a","project":"openstack/puppet-octavia","branch":"master","topic":"driver-agent","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-13 21:57:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07b12acbaba8eb8affc603c2679297495230d37a","subject":"Manage octavia-driver-agent package/service","status":"MERGED","created":"2021-05-04 08:51:55.000000000","updated":"2021-10-13 21:58:44.000000000","submitted":"2021-10-13 21:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789526-driver-agent","meta_rev_id":"60eec693662a9d7d33392f863b182f8c062dbebd","_number":789526,"virtual_id_number":789526,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 21:57:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-10-05 18:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-10-13 21:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-10-13 21:29:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 12:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 18:24:34.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-13 21:29:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"248a0a657476b1d1844ac1acc053accb20bcb0f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 08:51:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38908fa3659dec8ecd3727576614780be8cafe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 12:08:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f771bff2f1b48cfb1dad2ad8ef2d7e7 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80d53fd2f28a45ddb51b89f9af3c0501 : FAILURE in 4m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b63e7f9d9509419ab228dbf76a9b27d5 : FAILURE in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c318a5bd732b4f35b69642ecd91312d0 : FAILURE in 8m 56s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3e690aef50c40128de70a19aba8077f : FAILURE in 8m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c71242ca95e548f48cd1a97c444b3789 : FAILURE in 8m 58s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f06cb64d625456bad52b15e9e77ed0a : FAILURE in 7m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ddc1312fb3e47dd9791fb114e76e6d2 : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d83dfc75b31a4639a2a5d5c7330e8867 : SUCCESS in 27m 49s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e30c075172cd407688588e7f345b3ed6 : SUCCESS in 15m 05s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a867790e461a442b914b1204fef62434 : SUCCESS in 27m 27s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53b12e960074950a3a6bd3cdf4ce7e3 : SUCCESS in 14m 35s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ab16fb6a194480eb1a9eb778fa25afd : SUCCESS in 15m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964463af07e847509564e3270c075927 : SUCCESS in 5m 27s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44c8455fbd7947bf8098c5d90f41bab9 : TIMED_OUT in 3h 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9309254e8920317666c6fc8b9b5ea53ba7f7c204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-04 12:38:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48e0dca614836cba5fef22619a6acb4a4558f44d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:34:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6875e2199794c05912db0a596e98e1a : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4d2caf08f644c3a94c1199c6f508b1 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3100b99e0227415cbfd98df4c4cf65c2 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6442b8657e1a47f29ee4fb7654e7b23a : FAILURE in 9m 46s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd0e5ee7d130470f9b3dd5c6c5870a50 : FAILURE in 8m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4cf544772744204955794772e2eb28f : FAILURE in 7m 12s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7537fcb50e3418b8df2528b6f6f1777 : FAILURE in 6m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47c9419cdf4742f9ba35f974a8711b68 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/010caf1a2fb54ac7844547fc4dbc7934 : SUCCESS in 28m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3beb53536a0847549a00efbc1186434f : SUCCESS in 13m 49s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/482d94b6f31b4727812cb054676b4324 : SUCCESS in 29m 14s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5640bd4474d4d16ae5deb9e31515de0 : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24d6debeb4b4c79839f55958a4c33a8 : SUCCESS in 12m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb50120f34c344e385241719995f8c2d : SUCCESS in 4m 22s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5d3b3ee568c142b0a1a696e51bf0e1a5 : FAILURE in 2h 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6ef47657ca2392e94441a6b26bc6d17858d747c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-04 15:09:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b1dd3ad44d3fa59ca7ba31cca4d330ff68646429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 18:34:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17dd6ddd10dc46d3819f685c1b99fa0e : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/584c56a5ce40495b9fa0f90f1ede6f40 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557f727ef4004263bfd9034a7356bdb8 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f16ae8f5a8140e7b453d450a7384a27 : FAILURE in 10m 24s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcd9905d4b884c0faba4772fcc6864c3 : FAILURE in 11m 30s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb021892d6a440b4b1fc6d3e918ff57b : FAILURE in 8m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d175f6112c4a9cb942647808130eed : FAILURE in 8m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9be73d919a7a40a3821f48dcba1f0640 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33df069cbbc74dea80fbf977f1fe4a9d : FAILURE in 13m 53s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c45c424bf624db796e0fd9a40751987 : SUCCESS in 27m 36s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5828e3de9e324ad2a7e354805865353a : FAILURE in 15m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/680701f58fb7441daad6ad6ad1fb695c : FAILURE in 16m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/839b50500de44154aa24c95c85e2e7c0 : SUCCESS in 4m 17s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44b3e17ab8ae47b6983eeb940bfd1254 : TIMED_OUT in 3h 17m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b21bb1086409a689b6de565ef037708bc5375c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-04 23:36:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d8517188476321d14b1131433cb24ac4d64afd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 00:50:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7178f45b0d9f2de2a0f9279ebce9c8c93c38b835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 04:05:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/349f7cfd22fd465e8f2dc8791d83ed02 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1597a145c624fc590aacb2697bcfa24 : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000379291cb143f7bf11094a91d4b2ac : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c67a326771f43d19ebdfc3d54eef552 : FAILURE in 8m 51s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/77bed099cef24437be50400efc87930b : FAILURE in 10m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58e56e17f2d143b2acfbc9170e90403a : FAILURE in 6m 51s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/891d13724c714f369e579b25c6f358e3 : FAILURE in 8m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec025c46808a4b36a79f8eaf573d0bb4 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86a4b9a04e9c4914914ca8be87337914 : FAILURE in 11m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c62a5bebc2f0455888b88e9d2ff79556 : SUCCESS in 28m 23s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37bd8800b1724469b28219936b2d93c5 : FAILURE in 10m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4789d9c80ed942ca9015a98dd6baabf6 : FAILURE in 15m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b26476b0faf947de9d7f87f6a9d85885 : SUCCESS in 4m 22s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a26b20dee6a947bc8ee6f71047a902a5 : TIMED_OUT in 3h 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3f0533702c975945bb8d028976447a3fb4024552","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 04:51:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5bb5ccb82463e91ae5769ffc418bdad576b86acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 08:13:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d44836c1d396acd492d976e96288f1e2e4c714f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 11:32:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42c623f10b86432dbcbdab1eda153a9d : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c339fa0540804c2e8a9267e0f22eaacc : SUCCESS in 4m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8f4ab43fa2847dfa02b242201f14482 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b661911c3c7d461b842b7becb9605705 : FAILURE in 16m 02s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf7f90d14e1f41a2a17f527ea091c9e9 : FAILURE in 12m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bc9859006eb4f4c843826ec51e1ca86 : FAILURE in 10m 54s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22928b0a152544da9caf79f5a69f4e5c : FAILURE in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ba4d6ff17a44613937de18f483dedfb : FAILURE in 11m 00s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60ed38989da042dcb5a3a81cbf6b185f : FAILURE in 18m 13s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6d8cbe4051d24898b8ad635181d1d17c : SUCCESS in 28m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14d22cb0e6cd4dac86bbd817b9740923 : FAILURE in 14m 53s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75e807e7521645d9b451f3c3016b24e6 : FAILURE in 14m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a585ff3ab9214d00a86fcfe33f496753 : SUCCESS in 4m 31s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7f9f75ef4f0445bd969f34627f81612a : FAILURE in 3h 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a3f40929aa30f4f21e0f9bf80b4a2005b5571dbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 13:02:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2018a29212b634a4ec81e9315fdc06a876bf1839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 15:29:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d8426587c6244b38819e2918a9860b0 : SUCCESS in 7m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2318c616870e42c1b5d6d723fc75984c : SUCCESS in 4m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fcca99f8d18456dad2d74ec1816751e : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad25cb2b5ff6497f9558ab418481d7d5 : SUCCESS in 13m 00s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4f2b02a82bd461c8630b0cbc65fd4c8 : SUCCESS in 15m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6774ffe9cb974d0f8defac817a83887b : SUCCESS in 12m 05s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/463315b39b0d4243bcec7adb0634169a : SUCCESS in 10m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55edac1f59364b5ea4ca1f2fbe86719a : SUCCESS in 9m 02s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8810c30a2bc49c89edfda4aac1c9087 : FAILURE in 16m 59s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd050cbc9cf141338ac4bf523d0422d4 : SUCCESS in 27m 17s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/918b19b9caa0426c99451156b87505b9 : FAILURE in 16m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fe77b483bfe46e7b83d41d493af98b9 : FAILURE in 16m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37b70a9c6a2a442295ad2ba067503960 : SUCCESS in 4m 29s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec8b65cce9ee4f359d2a6569ea01337f : FAILURE in 2h 21m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"db520a0b0d0aa0fd647daa3343208e263e7a803c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-05 18:24:34.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2b5daded8a77bbe83898385c6a9315fbcb715568","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-10-13 21:29:46.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1e2136a78a8630eb40aa82489fec44fde7c9a05b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:30:59.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"98141a5ddbf789c66c79f863717cd5834b6a5a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:57:09.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69926b11640d448d87cf15c0dab169cf : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d99549923ee54e78b057b17cc831462e : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bbbc0777d614345a723c27189676f91 : SUCCESS in 3m 29s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/050d75d4563b4300ba3a72f0262b4b04 : SUCCESS in 10m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/27461b46da2845eb9b40879430c7df7a : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d224c01e955d42bfac44c93868eedfee : SUCCESS in 7m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/845ab0d1af414dfcbbc9276285c520bb : SUCCESS in 11m 39s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2ecbb8689964431b65fb6a8c7572e8f : SUCCESS in 25m 11s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/859e825dd3a84ca88dc87ae2e7c049c8 : FAILURE in 16m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/586a50f1098a4000b05bce18ed64ca49 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":8},{"id":"55d1c706f3acaa1625c02a4dd93b84d7444c4c23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"60eec693662a9d7d33392f863b182f8c062dbebd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 21:58:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/277674d675764e7fba2fc0ed670c8462 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"84b7c61586a49add0ad10cb4526593222236791f","revisions":{"e3d0cd9f57ab1cf4f7a9209a2f87a4f3ef3a1b32":{"kind":"REWORK","_number":1,"created":"2021-05-04 08:51:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/1"}}},"commit":{"parents":[{"commit":"6fba0ace08642dbc3e5f1f3e320d3f994cea6720","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6fba0ace08642dbc3e5f1f3e320d3f994cea6720"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:51:05.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e3d0cd9f57ab1cf4f7a9209a2f87a4f3ef3a1b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e3d0cd9f57ab1cf4f7a9209a2f87a4f3ef3a1b32"}]},"branch":"refs/heads/master"},"b4d56f96a825eee0c83f206745516985dd7d4534":{"kind":"REWORK","_number":2,"created":"2021-05-04 12:38:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/2"}}},"commit":{"parents":[{"commit":"6fba0ace08642dbc3e5f1f3e320d3f994cea6720","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/6fba0ace08642dbc3e5f1f3e320d3f994cea6720"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 12:38:11.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b4d56f96a825eee0c83f206745516985dd7d4534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b4d56f96a825eee0c83f206745516985dd7d4534"}]},"branch":"refs/heads/master"},"38726d1365a3704fd23815e4c2152d3cfba00212":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-04 15:09:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/3"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-04 15:06:15.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/38726d1365a3704fd23815e4c2152d3cfba00212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/38726d1365a3704fd23815e4c2152d3cfba00212"}]},"branch":"refs/heads/master"},"f96689981a20f4e45659c159bdba662d4884cd35":{"kind":"REWORK","_number":4,"created":"2021-10-04 23:36:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/4"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-04 23:36:09.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f96689981a20f4e45659c159bdba662d4884cd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f96689981a20f4e45659c159bdba662d4884cd35"}]},"branch":"refs/heads/master"},"28cf24e100c252956e27a703d73305ffc6971a71":{"kind":"REWORK","_number":5,"created":"2021-10-05 00:50:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/5"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 00:49:59.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/28cf24e100c252956e27a703d73305ffc6971a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/28cf24e100c252956e27a703d73305ffc6971a71"}]},"branch":"refs/heads/master"},"2120c178940d773b9d71f69100d74930a3834a40":{"kind":"REWORK","_number":6,"created":"2021-10-05 04:51:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/6"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 04:51:32.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2120c178940d773b9d71f69100d74930a3834a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2120c178940d773b9d71f69100d74930a3834a40"}]},"branch":"refs/heads/master"},"235a0267836f4ebf2c664d865263916c7bd11bcb":{"kind":"REWORK","_number":7,"created":"2021-10-05 08:13:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/7"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 08:13:16.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/235a0267836f4ebf2c664d865263916c7bd11bcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/235a0267836f4ebf2c664d865263916c7bd11bcb"}]},"branch":"refs/heads/master"},"84b7c61586a49add0ad10cb4526593222236791f":{"kind":"REWORK","_number":8,"created":"2021-10-05 13:02:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/789526/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/26/789526/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/26/789526/8"}}},"commit":{"parents":[{"commit":"f6b4783dc1badf69b349f155174a3306c2ff95a4","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f6b4783dc1badf69b349f155174a3306c2ff95a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-04 08:50:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 13:02:10.000000000","tz":540},"subject":"Manage octavia-driver-agent package/service","message":"Manage octavia-driver-agent package/service\n\n... since octavia-driver-agent is now packaged in Ubuntu and Debian.\n\nChange-Id: I07b12acbaba8eb8affc603c2679297495230d37a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/84b7c61586a49add0ad10cb4526593222236791f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/84b7c61586a49add0ad10cb4526593222236791f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
