)]}'
{"id":"openstack%2Fpuppet-octavia~790557","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~I38bb6932d280433174ea2139090f770cffba6fbc","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"typo-stable/wallaby","hashtags":[],"change_id":"I38bb6932d280433174ea2139090f770cffba6fbc","subject":"Fix a typo in parameter description","status":"MERGED","created":"2021-05-11 05:50:44.000000000","updated":"2021-05-11 09:39:52.000000000","submitted":"2021-05-11 09:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790557-typo-stable/wallaby","cherry_pick_of_change":790334,"cherry_pick_of_patch_set":1,"meta_rev_id":"d0fff63ab40b67dc5950ecd7561380ffc7382b36","_number":790557,"virtual_id_number":790557,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 09:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-05-11 05:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-05-11 05:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 05:51:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-05-11 05:51:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-11 09:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eae40d7f20503386b99c7009adf04b3dd0587cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 05:50:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2f468991f344fda9f1b8ee4caafbf366ffae7cef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-11 05:51:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI\u0027ll approve this by myself because this is a clean backport of nit typo fixing...","accounts_in_message":[],"_revision_number":1},{"id":"b20011e5f3cc17018c5375f1c45c2a34568932b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:11:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8404d2c5c9374cf7aa21bb699f2e4b81 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48bf7c4b0641462b9319d349f8c0d936 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e67092d1b49048b09f59a60ef682a4a9 : SUCCESS in 11m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/037e7b29800f463586b38d830ab293c1 : SUCCESS in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb128b8f5bc44bf6bfff972f4f570caf : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cd0d0b32e4049199fac52052b2d0e86 : SUCCESS in 24m 38s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/147c1f862b9a4b39bd53afba1aa9608e : SUCCESS in 12m 49s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ee3c436b1eeb4ed688f88da946c8611e : TIMED_OUT in 3h 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ff1b160e0c21e762531fdc2cc5066a6b8d97165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:13:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4caee61a1b915fcaa973559afbf03f10e58ec680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:39:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10f6706d8ab048a295ef501e386f3af9 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c679ff47344340d6ac9690ba07d1ae69 : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb6526ffde3f41849f328b6648c355b1 : SUCCESS in 10m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c769d433844140b9a6f3ab8a42ed0ddf : SUCCESS in 8m 42s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5cd5c33ea9de482ba9fbd85eaed79863 : SUCCESS in 23m 26s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10337a263b014728b9a1fe3cef2b6866 : SUCCESS in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0fff63ab40b67dc5950ecd7561380ffc7382b36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:39:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de98237632b52c6f0d998c300b8bb9d9cd35cd8f","revisions":{"de98237632b52c6f0d998c300b8bb9d9cd35cd8f":{"kind":"REWORK","_number":1,"created":"2021-05-11 05:50:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/790557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/57/790557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/790557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/790557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/57/790557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/57/790557/1"}}},"commit":{"parents":[{"commit":"55ce60902b04ebc0e219ae108f13762896afd307","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/55ce60902b04ebc0e219ae108f13762896afd307"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-08 14:31:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-11 05:50:44.000000000","tz":0},"subject":"Fix a typo in parameter description","message":"Fix a typo in parameter description\n\nChange-Id: I38bb6932d280433174ea2139090f770cffba6fbc\n(cherry picked from commit 3c37baefdc75efb708dfa8d4f46a9c699503104a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/de98237632b52c6f0d998c300b8bb9d9cd35cd8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/de98237632b52c6f0d998c300b8bb9d9cd35cd8f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
