)]}'
{"id":"openstack%2Fpuppet-octavia~792568","triplet_id":"openstack%2Fpuppet-octavia~master~I65aeab82eb4f62834be3162e9ce2bd3655bc315a","project":"openstack/puppet-octavia","branch":"master","topic":"pyvers-removal","hashtags":[],"change_id":"I65aeab82eb4f62834be3162e9ce2bd3655bc315a","subject":"Get rid of the $pyvers variable","status":"MERGED","created":"2021-05-21 10:10:00.000000000","updated":"2021-05-27 13:25:49.000000000","submitted":"2021-05-27 13:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792568-pyvers-removal","meta_rev_id":"b3e6eb04db07204e6da850a7f73bc092bf32df7f","_number":792568,"virtual_id_number":792568,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 13:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-05-27 12:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-05-21 19:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-22 15:27:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-05-27 12:47:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 13:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 19:32:48.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-22 15:27:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-05-27 12:47:51.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"0e296966e33bc431f2772a4df7a03b196d987a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-05-21 10:10:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e02fe99e8ba63f7a1f622c41b206f51328dd442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 13:33:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e92cdccdcce4d41bdb625bdc815268c : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7f9af09a1ca4b02a14db0b404fb6d4e : SUCCESS in 4m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b870e53b36d144c98c068b8408a9e137 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c8a7d8d6f294870b034834290bcec35 : SUCCESS in 11m 21s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d28601749ac649bb97af82960fb874fa : SUCCESS in 11m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be1ba5100f154afbbbd618a157b003ff : SUCCESS in 9m 44s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fe307a8d2da480d94af8bd09c2a95ac : SUCCESS in 10m 37s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f501f0ef05794ff4af68e09aa01e44f9 : SUCCESS in 9m 13s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/46c7a22013204b7aa0ca546d63dd5bb1 : SUCCESS in 32m 53s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24b592fb79ec493184f841b63782c0ff : FAILURE in 16m 02s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a8c7d6c87b74567b0967a766451e315 : SUCCESS in 24m 40s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b76966235d2b4f16a931faf51bedec60 : FAILURE in 12m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/351097852b8849f7862ebe2d9f299396 : FAILURE in 15m 12s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/52a1ca83a6144e31a093a2b90864bc1d : TIMED_OUT in 3h 15m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a9154a1e27b46e29099c355f5ae10f59334bc31","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-21 19:32:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc71433888ccd58ebf07c7b277c5cc45d120c652","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-05-22 15:27:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d1f6f4b3fa0cf1e9c87c87b65c18cd361d06331","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-27 12:47:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77e22880d3e4b8b470c05a7de18aa68964a87cd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f0fa235e8149aa26cbc4e9ffd10760ce8c70167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:25:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5ac1553a6214ec799485b98ed5f773c : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff877029876f498286ac3fc3f89af458 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d2282827634064803091c2762b3147 : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a24b774a1bb4ed8b6120619539c1ffb : SUCCESS in 12m 43s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/53ab7d74884d4f6c8a266cfe0517d3e9 : SUCCESS in 12m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1afd8b88cc0045338881de8f7fde6f64 : SUCCESS in 11m 38s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e882bd608ccf4783a9e65b3592a51b6a : SUCCESS in 10m 38s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/484165d4738f4047ba4749ccd9fca57f : SUCCESS in 31m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b20f1c871a84fd3ad59e06305d317f9 : FAILURE in 16m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3e6eb04db07204e6da850a7f73bc092bf32df7f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:25:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0221597f1bda5d82f28e6e320f0a0b2ee272c63f","revisions":{"0221597f1bda5d82f28e6e320f0a0b2ee272c63f":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:10:00.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/68/792568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/68/792568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/68/792568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/68/792568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/68/792568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/68/792568/1"}}},"commit":{"parents":[{"commit":"9448a989075bdf382690a25c6eddede2b047a51f","subject":"Merge \"Do not declare service resources when manage_service is false\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9448a989075bdf382690a25c6eddede2b047a51f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:09:49.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-05-21 10:09:49.000000000","tz":120},"subject":"Get rid of the $pyvers variable","message":"Get rid of the $pyvers variable\n\nSince everyone has switched to Python3, it\u0027s time for the removal of the\n$pyvers variable.\n\nChange-Id: I65aeab82eb4f62834be3162e9ce2bd3655bc315a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0221597f1bda5d82f28e6e320f0a0b2ee272c63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0221597f1bda5d82f28e6e320f0a0b2ee272c63f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
