)]}'
{"id":"openstack%2Fpuppet-octavia~796177","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fussuri~I1645afb89e33cfe03e25b09c446e2d47fc6b9f18","project":"openstack/puppet-octavia","branch":"stable/ussuri","topic":"bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I1645afb89e33cfe03e25b09c446e2d47fc6b9f18","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-13 02:48:22.000000000","updated":"2021-06-28 14:11:13.000000000","submitted":"2021-06-28 14:11:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796177-bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":794925,"cherry_pick_of_patch_set":1,"meta_rev_id":"6267ba81b411937bbebfbcf6a389cfad7bad3868","_number":796177,"virtual_id_number":796177,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 14:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-28 12:56:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-28 12:56:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-13 03:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea4d197d49baef628c79bb3d2e8add855c24aa2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:48:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"66a531f350a48ee76e766d01667f1ef03689c6d0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:59:37.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-stable/victoria-sluuvw5e7s removed","accounts_in_message":[],"_revision_number":1},{"id":"824e9c8cd8373b0558016106d59109e64528835c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 03:00:36.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"81e6ceb921069e782cd6df18787a6b6eef1b43cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 03:25:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0b4e7c3b9d843c3bb9f6709be760d45 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67d826fcedca446994eb2d93eda6fdb4 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a0edfd1aa3a45d7927eb63c09dff70c : SUCCESS in 3m 37s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d78cee0d1464305b13b06a163645655 : SUCCESS in 14m 17s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/503f57a643fc4e9496de707b508a3171 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/177c28efd8ef4c4c93ac296faee4e239 : SUCCESS in 8m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0953a67312234126b10abbcd1f199ec9 : SUCCESS in 9m 07s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9a6c10effdf74a7e984eae57f53d4aeb : FAILURE in 29m 25s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/1a31ab66c1d44a7d857f4ab2d4b1e869 : FAILURE in 29m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/004cb49c8d2744868667fe60d3621da1 : SUCCESS in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d8458cd607ed035532ca3e086707c34d5c680de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 12:56:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a6591b78f1df37373e71e5dd8a0f193132f7088","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:33:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fb26256268e428daa89ed5be0679f2a : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bc1bb80eedb44b3b6bcc0e6cd107e27 : SUCCESS in 6m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fddf60c974ba4cdca7973cd87d51d821 : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2339f8da6d34fee93cebd586d64ce68 : SUCCESS in 16m 55s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/7fcd975e40a64fbab5d7bcddd7cf1bae : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2353b15bddf94e8492954e9d2559e6fb : SUCCESS in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60cf7e6fa42c4d24bdeda75b5bfe99b5 : SUCCESS in 12m 44s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/43963fe61f4f4b9c8ca2b05a59101b5b : FAILURE in 30m 19s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f29392d3b334d66b496148944b64de3 : FAILURE in 33m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/999f0260612d4c4a96e97b7e42386b54 : SUCCESS in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"473284cabd3d4fe4098000baea986a372855f210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:34:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b4187d8306166c047560b61aba2675d9121acd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:11:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c71c4256769949efb0f2f4fdc9c31973 : SUCCESS in 6m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96af7a67a95e4defba8608f4b424463c : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b26aef246d241daaf20b58b6496b057 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/841f5504f4ee4085850e124991d42bcc : SUCCESS in 15m 49s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/5414bb0deae14511979f959ef7c64afc : SUCCESS in 17m 46s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7dc424a40234f21b00684ec613332f1 : SUCCESS in 10m 54s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c377eb439b7f4faa95ab3ddae4bdbf7d : FAILURE in 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6267ba81b411937bbebfbcf6a389cfad7bad3868","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:11:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee0b9f566647a84d85ec71a45e26730d8344b840","revisions":{"ee0b9f566647a84d85ec71a45e26730d8344b840":{"kind":"REWORK","_number":1,"created":"2021-06-13 02:48:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/796177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/77/796177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/796177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/796177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/796177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/77/796177/1"}}},"commit":{"parents":[{"commit":"bb73a687352ab7514e8defb80a31967f50c4db8c","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bb73a687352ab7514e8defb80a31967f50c4db8c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:10:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-13 02:48:22.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: I1645afb89e33cfe03e25b09c446e2d47fc6b9f18\n(cherry picked from commit 456e7ff27663471a27ff19be44637e6992572891)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ee0b9f566647a84d85ec71a45e26730d8344b840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ee0b9f566647a84d85ec71a45e26730d8344b840"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
