)]}'
{"id":"openstack%2Fpuppet-octavia~801314","triplet_id":"openstack%2Fpuppet-octavia~master~I6564e94550ea66f18fd99780a4c990fd18b4da4d","project":"openstack/puppet-octavia","branch":"master","topic":"listopts","hashtags":[],"change_id":"I6564e94550ea66f18fd99780a4c990fd18b4da4d","subject":"octavia-controller: Convert list values for ListOpts","status":"MERGED","created":"2021-07-19 14:30:46.000000000","updated":"2021-07-20 18:05:01.000000000","submitted":"2021-07-20 18:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801314-listopts","meta_rev_id":"63071b9de76d6c0e8e6d3801b9094cd06fd2b183","_number":801314,"virtual_id_number":801314,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 18:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-19 21:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-20 17:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-20 17:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 17:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 21:06:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-20 17:31:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d568c22707073cb49fb5b9bf103939fb6b132160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 14:30:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e12a165a7269ce393db7562a0038d8b911a1924d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 14:34:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9087e03da5a17d6f4f52e4e8f1a7e5243c5e23f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-19 14:39:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c382af928c5da6742508de08ae1ddd55ca484ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 17:35:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d08c8dfb5d3469280361e319df1cfff : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c32b12805a14912afb9059d4f3d36cc : SUCCESS in 4m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54967337bf8842b28a60116f1798fb22 : SUCCESS in 3m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebf26c11809646fdb3483768b3710917 : SUCCESS in 14m 20s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/068f93da0e9a410cb4357f11ad5beabe : SUCCESS in 11m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f413dd0648a844cab7c7a4ee923389a7 : SUCCESS in 8m 44s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d966bd70ee474c94bd5b300d6f541324 : SUCCESS in 8m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed9198106f14588a3cd9cdd5b930c50 : SUCCESS in 9m 08s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c1e56ef142d43c2805c59e801d2166f : SUCCESS in 31m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7b18f34722748f7aea95ccd68924511 : FAILURE in 14m 31s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5aefaf48ccd4968860275d763975f14 : SUCCESS in 35m 20s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/473d32b0396a4ab48c4b3ccdc616d91f : FAILURE in 14m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b10d5320fc39459e9d319348bdf31d90 : FAILURE in 13m 52s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d9f207c2bdfd43f88ce5def2c9791150 : FAILURE in 2h 50m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d10d71d8cbf03722340053ca261f539654e32f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-19 21:06:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c523dd9e0b6cfc480bad5c0f5e6229e0abb61428","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-07-20 17:31:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"44842af4bdd8150b931d7621a3b3d832af9a59e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 17:32:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0473ed7f65e01d8a22d242a03e50e46e19b554a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 18:05:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b44e15272f0548bfa9a2f09fa450676a : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cec7362c479c4f5a8ad2f05a67225aee : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e04cf37bafb043029b4e306b52141674 : SUCCESS in 5m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbb5ab09414e4dcdb9bf72a5b9528f5f : SUCCESS in 13m 11s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcfcbedcf980492082a1dbe426f42692 : SUCCESS in 14m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1027a069525b45aea7e83026eb3a8991 : SUCCESS in 10m 14s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3259135ee9104d93b8800861a2ad55e9 : SUCCESS in 11m 51s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f5887fe8ede4562b033fb077c5fd206 : SUCCESS in 31m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f1b6e5dd0d74f308a5f2c84d4fc48cd : FAILURE in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63071b9de76d6c0e8e6d3801b9094cd06fd2b183","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 18:05:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f62f0b0e15c4d26e377aa27e56aff5840c66a159","revisions":{"c4d8b2a2dcf3ab65a37cfd1a30bcaa207a35126e":{"kind":"REWORK","_number":1,"created":"2021-07-19 14:30:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/801314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/14/801314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/1"}}},"commit":{"parents":[{"commit":"a00e903ea85d13ccdd4a7038bdfc4f05c9798e12","subject":"Merge \"Add support for [api_settings] healthcheck_refresh_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a00e903ea85d13ccdd4a7038bdfc4f05c9798e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:27:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:29:49.000000000","tz":540},"subject":"Convert list values for ListOpts","message":"Convert list values for ListOpts\n\nThe following parameters in Octavia are defined as ListOpt which accept\ncomma-separated list. This change ensures that the corresponding\nparameters in puppet-octavia accepts list values and convert these\nvalues to strings in the proper format.\n - controller_worker/amp_secgroup_list\n - controller_worker/amp_boot_network_list\n - health_manager/controller_ip_port_list\n\nChange-Id: I6564e94550ea66f18fd99780a4c990fd18b4da4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c4d8b2a2dcf3ab65a37cfd1a30bcaa207a35126e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c4d8b2a2dcf3ab65a37cfd1a30bcaa207a35126e"}]},"branch":"refs/heads/master"},"63b180cfb82f2450f79dfb13420f50a4f89e70dd":{"kind":"REWORK","_number":2,"created":"2021-07-19 14:34:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/801314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/14/801314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/2"}}},"commit":{"parents":[{"commit":"a00e903ea85d13ccdd4a7038bdfc4f05c9798e12","subject":"Merge \"Add support for [api_settings] healthcheck_refresh_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a00e903ea85d13ccdd4a7038bdfc4f05c9798e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:27:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:34:02.000000000","tz":540},"subject":"Convert list values for ListOpts","message":"Convert list values for ListOpts\n\nThe following parameters in Octavia are defined as ListOpt which accept\ncomma-separated list. This change ensures that the corresponding\nparameters in puppet-octavia accepts list values and convert these\nvalues to strings in the proper format.\n - controller_worker/amp_secgroup_list\n - controller_worker/amp_boot_network_list\n - health_manager/controller_ip_port_list\n\nChange-Id: I6564e94550ea66f18fd99780a4c990fd18b4da4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/63b180cfb82f2450f79dfb13420f50a4f89e70dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/63b180cfb82f2450f79dfb13420f50a4f89e70dd"}]},"branch":"refs/heads/master"},"f62f0b0e15c4d26e377aa27e56aff5840c66a159":{"kind":"REWORK","_number":3,"created":"2021-07-19 14:39:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/801314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/14/801314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/14/801314/3"}}},"commit":{"parents":[{"commit":"a00e903ea85d13ccdd4a7038bdfc4f05c9798e12","subject":"Merge \"Add support for [api_settings] healthcheck_refresh_interval\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a00e903ea85d13ccdd4a7038bdfc4f05c9798e12"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:27:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 14:38:13.000000000","tz":540},"subject":"octavia-controller: Convert list values for ListOpts","message":"octavia-controller: Convert list values for ListOpts\n\nThe following parameters in octavia-controller are defined as ListOpt\nwhich accept comma-separated list.\nThis change ensures that the corresponding parameters in puppet-octavia\naccepts list values and convert these values to strings in the proper\nformat.\n - controller_worker/amp_secgroup_list\n - controller_worker/amp_boot_network_list\n - health_manager/controller_ip_port_list\n - amphora_agent/admin_log_targets\n - amphora_agent/tenant_log_targets\n\nChange-Id: I6564e94550ea66f18fd99780a4c990fd18b4da4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f62f0b0e15c4d26e377aa27e56aff5840c66a159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f62f0b0e15c4d26e377aa27e56aff5840c66a159"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
