)]}'
{"id":"openstack%2Fpuppet-octavia~803451","triplet_id":"openstack%2Fpuppet-octavia~master~I1545d1e02c0b5290303a2399c881be032f02fbc2","project":"openstack/puppet-octavia","branch":"master","topic":"remove-tripleo","hashtags":["puppet-tripleo-job"],"change_id":"I1545d1e02c0b5290303a2399c881be032f02fbc2","subject":"Remove broken TripleO job","status":"ABANDONED","created":"2021-08-04 09:20:29.000000000","updated":"2021-09-16 11:10:21.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"67bfc876cc0299e87af5cbe1eb14e0bc7c4273cc","_number":803451,"virtual_id_number":803451,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-04 10:03:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-08-09 07:14:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":-1,"date":"2021-08-06 07:35:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-04 09:52:30.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-08-04 09:53:00.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2021-08-04 10:03:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f5d3813a17fda4b43c9cef3cc2053995bd8416a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 09:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3eef5ebc2b0875f2c6d6fbdff291697c40e3432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 09:26:09.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"33bdf2c7ed38e378a2b121ed12d86b24ce81f119","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 09:27:09.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":2},{"id":"a5296d2e223f8f926a1e065125fc8e261c52d7d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-04 09:31:20.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/c/openstack/puppet-octavia/+/801400 might be able to fix the job (the first trial failed. rechecking the fresh result)","accounts_in_message":[],"_revision_number":2},{"id":"0f8b56a5a9acdd98b76c1ea46af3f1f9b41beb5b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 09:52:30.000000000","message":"Patch Set 2: Code-Review+2\n\nagree this should be removed now\n\nif the job is wanted/useful then we can re-add it once it becomes stable","accounts_in_message":[],"_revision_number":2},{"id":"7f903d741e7410ff9f7d3c99799b8ceb1736f3d4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-04 09:53:00.000000000","message":"Patch Set 2:\n\nadding beagles i think he works on octavia/scen10 stuff FYI","accounts_in_message":[],"_revision_number":2},{"id":"f6e44a37fb92f26880b4ecd042c82bb8a9c4115a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-04 10:03:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b9d21717ef641d19e94fa0a46a27d78 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1efd39afe1524affa75ac79444c57d20 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/210346182fcf4e538adac74276f931b8 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0edffdae6bee4e4aa7db34f882e9792d : SUCCESS in 11m 59s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/67d93e8461a840b49ea493283ec9bb09 : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97726a68597b455d95bf9c1d7a56b079 : SUCCESS in 9m 30s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c2172e61b0b43bca850406af790b258 : SUCCESS in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60890ebc87584f388e4b40963fe7cf80 : SUCCESS in 9m 34s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2042a6852dd47fa8b98250492e526b0 : SUCCESS in 28m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/261c9a11795b4a318c069cec53d98e93 : FAILURE in 13m 06s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/65b6f6c2a448447589f173ba9c6e9eb5 : SUCCESS in 31m 07s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caf9642a401a45689eda9d2ffa7b526f : FAILURE in 13m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc88c7a377fc4c81984d6829528c92b8 : FAILURE in 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b585177a904b2156143c3a2168ea94ea5b055c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-06 07:35:57.000000000","message":"Patch Set 2: Workflow-1\n\nThanks to Michael, https://review.opendev.org/c/openstack/puppet-octavia/+/801400 is becoming primising now so let me hold this until we get some conclusion there.","accounts_in_message":[],"_revision_number":2},{"id":"0aa05dce3a51297a49b785f7bdd50afe0acbd039","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-08-09 07:14:03.000000000","message":"Patch Set 2: -Code-Review\n\nack removing vote then thanks","accounts_in_message":[],"_revision_number":2},{"id":"67bfc876cc0299e87af5cbe1eb14e0bc7c4273cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-16 11:10:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7db8176990e00da576dcf85d445c9fac50f683d5","revisions":{"810cc52e51858a86ec6bb5a714f143b8534fc56a":{"kind":"REWORK","_number":1,"created":"2021-08-04 09:20:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/803451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/803451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/1"}}},"commit":{"parents":[{"commit":"ff21d5893ed5c83073f22bf14ced1b2c36ef9af5","subject":"Merge \"Add support for MultiStrOpt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff21d5893ed5c83073f22bf14ced1b2c36ef9af5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 09:18:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 09:20:15.000000000","tz":540},"subject":"Remove broken TripleO job","message":"Remove broken TripleO job\n\nThe puppet-octavia-tripleo-standalone job (or a similar tripleo job)\nwas added to this repo a while ago but this job has been broken and\nnot maintained. The job is continuously failing and consuming a lot\nof resource and time (usually more than 3 hours)[1].\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dpuppet-octavia-tripleo-standalone\nChange-Id: I1545d1e02c0b5290303a2399c881be032f02fbc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/810cc52e51858a86ec6bb5a714f143b8534fc56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/810cc52e51858a86ec6bb5a714f143b8534fc56a"}]},"branch":"refs/heads/master"},"7db8176990e00da576dcf85d445c9fac50f683d5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-04 09:26:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/803451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/803451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/803451/2"}}},"commit":{"parents":[{"commit":"ff21d5893ed5c83073f22bf14ced1b2c36ef9af5","subject":"Merge \"Add support for MultiStrOpt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff21d5893ed5c83073f22bf14ced1b2c36ef9af5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 09:18:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-04 09:25:24.000000000","tz":540},"subject":"Remove broken TripleO job","message":"Remove broken TripleO job\n\nThe puppet-octavia-tripleo-standalone job (or a similar tripleo job)\nwas added to this repo a while ago but this job has been broken and\nnot maintained. The job is continuously failing, wasting resource and,\nwhat is worse, consuming very long time time (usually 3+ hours)[1].\nBecause we\u0027ve seen no effort to fix that job, let\u0027s remove the job\nto speed up CI jobs.\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dpuppet-octavia-tripleo-standalone\nChange-Id: I1545d1e02c0b5290303a2399c881be032f02fbc2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7db8176990e00da576dcf85d445c9fac50f683d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7db8176990e00da576dcf85d445c9fac50f683d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
