)]}'
{"id":"openstack%2Fpuppet-octavia~804173","triplet_id":"openstack%2Fpuppet-octavia~master~I097df997bbe66f274c76a29d42f49ffdba2c9575","project":"openstack/puppet-octavia","branch":"master","topic":"quota-cleanup","hashtags":[],"change_id":"I097df997bbe66f274c76a29d42f49ffdba2c9575","subject":"Drop quota parameters without default_ prefix","status":"MERGED","created":"2021-08-10 23:32:30.000000000","updated":"2021-08-12 18:47:14.000000000","submitted":"2021-08-12 18:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804173-quota-cleanup","meta_rev_id":"7159ddc5310f6d38d2ec216eff49333fe6cb0c4f","_number":804173,"virtual_id_number":804173,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 18:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-08-11 06:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-08-12 18:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2021-08-11 06:37:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-08-12 18:16:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 02:46:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 06:37:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-08-12 18:16:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5dc991083bf90ff3149033e1a51feb7f9c9cee5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-10 23:32:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2b97eaee5a0f696e827f29a00bbc99c6c05256c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 02:46:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aa87e258e3a456cb27339f7c8eb70d0 : FAILURE in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38d1c5167f2843a89252196c85659efb : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac66b3daaa0c4ec4a92d2f2df242df82 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1ae0f82242c441cb9dc948d891d717e : FAILURE in 12m 18s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/98b2d9cdc722492786cb6e8b99542e6c : FAILURE in 12m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b2873c85ce346ad8c7d35dd256e667d : FAILURE in 8m 57s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b708ee9abbc41c19b155cff5ffd66b1 : FAILURE in 8m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e3d8c385f3641fd83f3dac300dbc3ff : FAILURE in 9m 28s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f51fad783f442c59411e0c319aa7d77 : SUCCESS in 26m 10s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55d8266be9034769a0860f14db5233c4 : FAILURE in 14m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/761813e63f8c44158ad2be205f4e5cff : SUCCESS in 26m 34s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0af9bf42fe4005b7eb31532b482fcb : FAILURE in 14m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7f95f0b77274281b5def18d10b06763 : FAILURE in 29m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/252d2934ead3425199f119f27d435dc8 : SUCCESS in 4m 15s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eb856eaf5c6e4337affdf34b943e6436 : TIMED_OUT in 3h 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd69faa501332b8c4cd5a327b39d96f1960645bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-11 02:54:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"98089ed517cff17eab801654fc35ee3b5146585b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 06:14:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b84326e57b844868a8e881bbee06669 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4528345c008c44458b9effad138ea816 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d04e976b4874ed5a9427b02a713754f : SUCCESS in 4m 19s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/792a754f2771451999426c5bdacd2c3f : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/57a2a9ea40ed441898d12f796ff417fc : SUCCESS in 11m 06s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa8d4ba296e7428dafd3abab3db7c35b : SUCCESS in 7m 35s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3306f1ea5204aceab2e2dc113d02e7e : SUCCESS in 8m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c7b4f322804423cb802d6fe9cae3099 : SUCCESS in 8m 38s (non-voting)\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/502e603d321b45ae929ee5285131e228 : SUCCESS in 23m 53s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54069120dac7451687e933fcc4e0b441 : FAILURE in 14m 09s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/608b6c858ac940cab3484dae8bf0156d : SUCCESS in 22m 25s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b21823867064e39ac6ad82d921c4b1d : FAILURE in 12m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c5a6778e7344812918cea92cb30199b : FAILURE in 14m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f828b2350004a88b5a2e2e157175589 : SUCCESS in 3m 44s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/57609429a603477687460ab2dc183742 : TIMED_OUT in 3h 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fad530937dbca25e3d8f6bcfc3ba2986b9087aa9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-11 06:37:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5185218ae126217a30431b964b231d4037048135","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-11 06:37:42.000000000","message":"Removed Workflow+1 by Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"4807be220920499fee402c9b3d5deff4827c805c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-08-12 18:16:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0ad857e3f0d63a8018225df5872fba46f6fe4e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:17:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92386652e6d3b0ec2b2236206189b2073c09a9f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:45:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5110831c1bb145e6bb9a00bfff449881 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d83485ceb86f40dda31530dc8eb01705 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b2292cba65a4b849fb3eb2ca4e84d7c : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed3a7727c05a45e788577858639d3f66 : SUCCESS in 12m 58s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f602ea859e24936acc19f0ea13139c9 : SUCCESS in 13m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79151423a01e43f5b03f90c983cfbfcd : SUCCESS in 10m 37s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a7956a430a84898b64f0c30cf3e469f : SUCCESS in 10m 42s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b20ade1fd5ad4dceacb8d0e0fbf8b1e3 : SUCCESS in 24m 59s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ee331e745da45c0947af1ce7e5dc3e2 : FAILURE in 12m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b8e336081454a4d8ad5072b07acdabb : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ce70ecf409af2a80cda241abbd05e7af5128923f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:45:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7159ddc5310f6d38d2ec216eff49333fe6cb0c4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 18:47:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e0c5958600b49c0bebbe9eb7134ca2a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"12d06f3636b4c95f1e40e4cb7cd195bb3eafbbb6","revisions":{"4b90dc4e665f8a3effabdfcb5f18bfd3fc175d90":{"kind":"REWORK","_number":1,"created":"2021-08-10 23:32:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/804173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/73/804173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/1"}}},"commit":{"parents":[{"commit":"53f69cd9b65d215fe1c2d7dd7b7d3d31cc69aac3","subject":"Prepare Xena M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/53f69cd9b65d215fe1c2d7dd7b7d3d31cc69aac3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-10 23:31:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-10 23:31:38.000000000","tz":540},"subject":"Drop quota parameters without default_ prefix","message":"Drop quota parameters without default_ prefix\n\n... because these parameters were deprecated during Wallaby.\n\nChange-Id: I097df997bbe66f274c76a29d42f49ffdba2c9575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4b90dc4e665f8a3effabdfcb5f18bfd3fc175d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4b90dc4e665f8a3effabdfcb5f18bfd3fc175d90"}]},"branch":"refs/heads/master"},"12d06f3636b4c95f1e40e4cb7cd195bb3eafbbb6":{"kind":"REWORK","_number":2,"created":"2021-08-11 02:54:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/804173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/73/804173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/73/804173/2"}}},"commit":{"parents":[{"commit":"53f69cd9b65d215fe1c2d7dd7b7d3d31cc69aac3","subject":"Prepare Xena M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/53f69cd9b65d215fe1c2d7dd7b7d3d31cc69aac3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-10 23:31:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-11 02:54:36.000000000","tz":0},"subject":"Drop quota parameters without default_ prefix","message":"Drop quota parameters without default_ prefix\n\n... because these parameters were deprecated during Wallaby.\n\nChange-Id: I097df997bbe66f274c76a29d42f49ffdba2c9575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/12d06f3636b4c95f1e40e4cb7cd195bb3eafbbb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/12d06f3636b4c95f1e40e4cb7cd195bb3eafbbb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
