)]}'
{"id":"openstack%2Fpuppet-octavia~805517","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~I3d879b2b8dd6110dfc1949d377c36c9cc3686ed7","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"I3d879b2b8dd6110dfc1949d377c36c9cc3686ed7","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-08-22 05:11:43.000000000","updated":"2021-08-22 09:06:08.000000000","submitted":"2021-08-22 09:06:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805517-tripleo-tempest","cherry_pick_of_change":801400,"cherry_pick_of_patch_set":7,"meta_rev_id":"025186412175e0883340c373198b365d974a9036","_number":805517,"virtual_id_number":805517,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-22 09:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-08-22 05:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-08-22 05:12:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 08:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"54af1cb9ad2d59456ad393690f77b19db261fa1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:11:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b95cd9ff5d94385a9c51036296552481990ecb0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:12:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9567cbb672202321d9798588c09fe0af54b21965","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:13:10.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"4d3030e6920397b36f8dafca1b5c06d16e8d60d3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:13:12.000000000","message":"Topic tripleo-tempest-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"10323330237ba2835d6b67be83856064b7c09057","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:13:16.000000000","message":"Topic set to tripleo-tempest","accounts_in_message":[],"_revision_number":1},{"id":"81e4933ffc1348a105b8142b66acebf22b8ba159","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 08:35:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b03f58105a904589a3d520dd177f618c : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd94da29b686439db1e5c311bea1f543 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8da429a05134f8bbf7868a4eac187f2 : SUCCESS in 14m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78fa673c483345f9814d5c2df466dfb6 : SUCCESS in 9m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/318e1e52754242dab856ebf8d4a6d1fa : SUCCESS in 8m 59s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/42f10632da5d42fb9b6f72aa22933cba : SUCCESS in 24m 52s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9c53197231d4b51900a8dc3e9d6d31c : FAILURE in 14m 28s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/363f5a51a7e64dc686e99fa00cc4b850 : TIMED_OUT in 3h 18m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e7b3e56709a82d75f5375e965e4df1beae7c4a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 08:36:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7ec689dfc29007f23d02e17010a30538633d286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 09:06:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d30a2a56eca0433a8a3982c004fe856f : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77e87241c15a480699359e909847188b : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9294419c65b4a0ab92567bd77316fda : SUCCESS in 11m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa10f274477d4e9bad6c35bde7e1b64a : SUCCESS in 10m 56s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/bce17ec0a74d450e95623ddb10f41fe3 : SUCCESS in 29m 13s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b37c7b5f50224600818c90a57ebc74dd : FAILURE in 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"025186412175e0883340c373198b365d974a9036","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 09:06:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7658c66ad36072fa3c265ce5d0f95cf1510cbb68","revisions":{"7658c66ad36072fa3c265ce5d0f95cf1510cbb68":{"kind":"REWORK","_number":1,"created":"2021-08-22 05:11:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/805517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/17/805517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/805517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/805517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/805517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/17/805517/1"}}},"commit":{"parents":[{"commit":"7f1264d7c205a8639402ea964f3eebf15ef10b67","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7f1264d7c205a8639402ea964f3eebf15ef10b67"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-19 23:25:55.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-22 05:11:43.000000000","tz":0},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests executed in TripleO job\nso that we run the same set of Octavia tests as the ones used in\ntripleo-ci-centos-N-scenario010-standalone job in TripleO repos which\nenables Octavia with Amphora driver.\n\nChange-Id: I3d879b2b8dd6110dfc1949d377c36c9cc3686ed7\n(cherry picked from commit 5565bb60a0a3de5263a0f8bbeb701305c0696e82)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7658c66ad36072fa3c265ce5d0f95cf1510cbb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7658c66ad36072fa3c265ce5d0f95cf1510cbb68"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
