)]}'
{"id":"openstack%2Fpuppet-octavia~816434","triplet_id":"openstack%2Fpuppet-octavia~master~I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd","project":"openstack/puppet-octavia","branch":"master","topic":"octavia_jobboard","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-08 00:52:37.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-03-08 00:52:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd","subject":"Add parameters to configure amphorav2 driver","status":"MERGED","created":"2021-11-03 07:15:21.000000000","updated":"2022-03-08 00:53:55.000000000","submitted":"2022-03-08 00:52:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816434-octavia_jobboard","meta_rev_id":"adfa9817ea859909abe62e14973b9395718fd2fd","_number":816434,"virtual_id_number":816434,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 00:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-03-04 04:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-03-07 23:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-03-07 23:40:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 09:54:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-11-03 10:35:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 01:23:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-12-08 21:20:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-14 19:51:47.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"}],"messages":[{"id":"dc4197d9491401c8bef216fa304212fae27f9f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-03 07:15:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d49c932876e7dcb9b2a5eb56ea3da0a34ca0ae71","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-11-03 07:15:32.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"97d8f5bb1d69a619e930f90a5722144c04790071","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-03 09:54:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd836cd5efba5e938a6ad5f4bd5d02a85f475577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 10:35:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21f3176f91524fa7a5afef7d5af9ccf3 : FAILURE in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988f34fffddf404f989e529c4681f7ea : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3d068e6c11b489199af97a12594d521 : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfa2ce3b30614c5cab799a36fe8fcce6 : SUCCESS in 13m 56s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c769e138bedd4ad7a421b50da9269ddc : SUCCESS in 13m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b786ad313fee4bb19c08915ba3b35d5e : SUCCESS in 11m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/383554f82a9d4935a6d87ee1e871e731 : SUCCESS in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/145ac9ba02b747fbbca779a3189c4e25 : SUCCESS in 9m 24s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6caad8ecd92146888495357b4d8180fa : FAILURE in 15m 19s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/81969c46773b4e87b1a7a9281cfd2584 : SUCCESS in 29m 50s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/342c159b28944ce5b48b3faf30cc9ea4 : FAILURE in 14m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c1fcd311b724dc583f8757e401ec463 : FAILURE in 16m 08s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a621befebd7b4a13a0de22cc94910b0d : TIMED_OUT in 3h 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"999e5f38a1f6ff400fb58fe8a46893d165b4a518","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-09 15:49:08.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fdc0ce3e3335e1be76bee8b210d12534aaabf1ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 14:50:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a490f29e256675b801c6a15029e21543199d691d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:55:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/845755c8f12f4322ba7ad16a72bc06ba : FAILURE in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b0488e22a9044f0bef59ccfbb49a266 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23e278d01781409590b3db8199b5c278 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fb760f51384441f8417943e63a2967a : SUCCESS in 14m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/05cb1701f4fd4df3b8d3afa9d21cea9c : SUCCESS in 12m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/752be474dcf34642bd69dcd2ee5874e0 : SUCCESS in 9m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdad6b81e22a43bd8cacc158e1e6d6ff : SUCCESS in 15m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34548bb9829c42a88b723dd33b9eef65 : SUCCESS in 12m 09s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640cf4612dd94571a59daf75d242d7fe : FAILURE in 12m 40s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/29ee75e4d1ba4fb2a3c5ff86d5d1508e : SUCCESS in 28m 39s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d28025beb044927816c2ce0e561a79f : FAILURE in 15m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b2d85d3979e4be39f29fa68af403944 : FAILURE in 18m 16s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ca6ed838b864cadb8d16892aa3d4e91 : FAILURE in 1h 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4fd1c27dfa46d313f7b8a5766c3a3c0723b0f788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 17:02:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3da21356ebe5f1b6b1a34487d632b109c164456c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-06 17:07:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"62719ae1d16ba4b0aba9104c4d817b16e1404386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 20:19:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13ec8a3a230b469db8d405714d250de4 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e78ffb397fee4ea9a6e3511555af48ac : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a98e6540b0a74e4487d0b87af28ce50a : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f51062f20444a0db6df687b288ad5a6 : SUCCESS in 14m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7327b6922298432491968ac1e2865879 : SUCCESS in 13m 55s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/486fa165198c4d40a8cc9bf13478555f : SUCCESS in 10m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccf56f5ae1764fd49ad900b3e4b37a23 : SUCCESS in 8m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b97964c9abd4fa6b86734005d66785f : SUCCESS in 9m 46s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73d9e9597be64a6799b932d8bbb4c003 : FAILURE in 13m 58s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/613af0b110ec47a5aaa76539e93c32b2 : SUCCESS in 30m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7da2b75529e44540963dbf5cb6c86b76 : FAILURE in 13m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d50d32f809d46ceafb7430368b9fa66 : FAILURE in 12m 54s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/455d97384cac486ba9b4a01942c98cf9 : FAILURE in 3h 03m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b8366b0b353c81d0f63ea8505ae7468dbcece005","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 01:23:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"eef2584f5369bb9f0f029928afdbbbd1eae5b945","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 01:25:42.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b056937b03730e35254aca759427c57afeb96ea5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 01:54:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9eead2a29ecbe71f82d5ed4ee35463d92cc1b24e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 01:55:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b92d7adf89a571a856714155e585ffcecc4469ec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 01:57:37.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"67c2f4ec90210a95603774fb238ab614b81d48b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 02:40:59.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2c8d5968ef6a534c65f58221d71b6a62ef575db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 02:57:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6b888132a9c7b13e27803d8d3f425166be77f5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 04:08:23.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"71a5e693dbbfdf7b88a2b036ad509f320d7d2e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 06:47:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d6493ff7110413094dbfc06d737f77b : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/592be27473ee40d49e9fe8611fdb50ae : SUCCESS in 3m 27s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff9b371bf27a48a28ed04d4e04768142 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dab16dee4a66453a92e55f71cae18026 : SUCCESS in 11m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e1e0f14bbac4e8d849c97100b15990e : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9137842998c4c6e9aa1c71c867dff76 : SUCCESS in 8m 49s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a41963cb0ca479385b4accf3323a77b : SUCCESS in 9m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11e596095bdd4cc1b06aaf921ba16980 : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf7f830bc06e46018450d35251f6d87c : FAILURE in 12m 44s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/700b3f82d6a44c15a5b15f1356183136 : SUCCESS in 26m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2811e21b4387495d97cb8e1ff1e37d99 : FAILURE in 13m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/446f57f5f4ac40299432a5078cdc9081 : FAILURE in 12m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/858e5c5c75164b0eae35b98e308e219e : SUCCESS in 3m 51s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c8825488392846689cd3468e41abfdb6 : FAILURE in 2h 32m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"85a07b8c406202d7f9983e5b048648cadf753544","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-07 07:40:44.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a85fe4ee8e41534ec4c59b8acbfb308f2c98630d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 07:47:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fe6e6899c87790386f5788bae8775fbc16255466","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-08 21:20:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0a902e5f7dcc885b42df03316baf4b7900eb2bf3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-12-10 13:07:21.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3a803b7a25fa01c78419b157dbb2ab74dc85e3fd","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-12-14 19:51:47.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"8e0123fbe6fd86b6619504202cd37070cf5e3b60","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2022-01-25 18:31:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"09f90ddf3d140fcc7b4c9d54fe30168149486bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 21:22:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d10bcd25b8dc4ec3a50c3afdd41086a9 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f1e9ecb446747c2a5da2b8e1ef8cbd9 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92182331487b4b2fa091ff154e0f24a2 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b76740dd22744308a36706e09b7bf28f : SUCCESS in 11m 52s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/842f322c11154cc999e5ec219f096828 : SUCCESS in 12m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e49adea868e64cafbb61f95cfd0b2447 : SUCCESS in 11m 06s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd1d3aee783740569ea201ab9db8e818 : SUCCESS in 8m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f44fc4e855d486988ad2539a343d28c : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2edac00c187748f9a500613b31c21aba : FAILURE in 13m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d67bbbbac83f46b4bfee097f2720fe99 : SUCCESS in 30m 00s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6157e8088a044b6b6c4c3198f253099 : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/425e838b972b4ee4842342dc10ae6588 : FAILURE in 15m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/145d65e210914b618076d5e03b72263d : SUCCESS in 4m 43s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/88f0030f306b4ec39ac11a162e579410 : SUCCESS in 2h 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4742f77208c000134e1eb00cc901418ed41890e2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-28 11:01:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4160be9ae9f69f66b43df71996cae97a644be3cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-28 14:40:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"465cb937a3cb73eaddd7b74884c42932b4921765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-28 14:44:29.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"db3752d2d595fa42fad2fe5546eb0d5d00ec24dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 16:48:46.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ab99236760d4a539f73e6292cabb480 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a8c605a8cc64b16bad782a8869d3b77 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23582ebcf4454bc5aa70080eab20f25d : SUCCESS in 5m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb574dec361e4dd686931cf05d5fd852 : FAILURE in 12m 48s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c56423e8eb042dcb125c19ac60d2006 : FAILURE in 11m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a812bdc71274733a87574e2a137bba4 : FAILURE in 8m 50s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b0cb27a00ad43cd9a99fc7f3088b305 : FAILURE in 9m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b633855215bd479cb0e77ee0787a445c : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf99750256ae4c2eb0db98b30a7d160e : FAILURE in 28m 30s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2c7ac054c2c42eea23da25a96ec5501 : SUCCESS in 57m 00s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc685987257f4492a329901893ef0c1e : FAILURE in 46m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db66a2ec1fba4b62959a26011e7a368a : SUCCESS in 19m 50s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c49002effe5344d2a8775bfca9b9034b : FAILURE in 38m 41s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/692a21735a054ccaafac0708ffa94893 : FAILURE in 10m 00s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84fd8e278e574dbe8e9f33cc40892e0b : SUCCESS in 20m 14s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5da41261ce2488e93a1fbd08c61f5b1 : SUCCESS in 20m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71772eb75b12476a9acbf58ea67101c3 : SUCCESS in 4m 48s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/896e195fac05467eb6fd80aeff323cfd : SUCCESS in 2h 00m 24s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1e535ef25afa8aee2f5a1974a7b60c0698660c18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-02-28 16:52:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1ff015f0b144b8baa3e7f039996c7383574b2783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-28 23:04:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e1ac81006ffb503f9a6726502acbf078a62d72f2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-28 23:06:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"072d1f64d297d408dd1dc60cf5fa58c1e195ff38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 01:06:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea23a1138bb2495abb8e34599178a2cf : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920713782064448a9e05c53f0b7f47bb : SUCCESS in 4m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816d874cfcd64270a0b61a1345cb9da0 : SUCCESS in 5m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e076badb1ac440498a534ba0d507c40 : SUCCESS in 15m 53s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e045274246142af8543b2099e6e98f8 : SUCCESS in 13m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a9ca8f4d02d4b10a6e75599fcb188d2 : SUCCESS in 12m 19s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d548839efb924df5b4b3d6e22a41b5e1 : SUCCESS in 11m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fa14c8358474e75a27d434e58121451 : SUCCESS in 11m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a8f6effb90842c8af3cc96ce7f5dfc6 : FAILURE in 30m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b82b92d5bf514835a69b6f5db0623de8 : SUCCESS in 58m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4573622158f4898aefa83c37445ffe8 : FAILURE in 47m 06s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57456f765b0640c39c903d72ab048c7f : SUCCESS in 21m 51s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7bff5030c95e4d0794fc8adc3c48aad8 : FAILURE in 36m 24s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/70be5e6b1a644c689c5b44db2726d940 : FAILURE in 8m 46s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07a2dcd7355641bdbe3dc0dcc1d3df31 : SUCCESS in 18m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8f6a2fb3a2040a3bd6896079359277f : SUCCESS in 20m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c932699dfce4706b8444f848eabe343 : SUCCESS in 4m 55s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/83458518314c47dd84315530b1a5bedf : SUCCESS in 1h 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2582d0b32431d6688023483fcb6cd20ef66652b5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-03-01 07:38:44.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"02fe127391cfd950c35b4353f52ca7b042d04e97","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-01 15:25:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3aa554241588207f9c0fa94a434f90f3af12c1b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:31:51.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7674d717636a4d058af041cb0b2a97d8 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e22f6ed6b49d4f7a8434194e27ef28cb : SUCCESS in 6m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebd80083729e474bb7a873b7a269d34c : SUCCESS in 6m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc75ff63ce324d24b104e5bd825b891e : SUCCESS in 19m 18s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1d6c01092ea4787a8422ff8bfbc38ed : SUCCESS in 17m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/235fda0af90444f7b7a36c9f892e9947 : SUCCESS in 13m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c1d880292b4a87adbbb23735c4f64d : SUCCESS in 13m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e76d65eff7742f6a98e6519dbd5179d : SUCCESS in 13m 37s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfd864ffdca24b61b339360a2a7b5a06 : FAILURE in 43m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc711c3938c64f32ab649b56f3f3946f : SUCCESS in 1h 01m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c35826a90c642d8865b43438fb63f4d : FAILURE in 48m 09s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/079640d2fd9c41d2a42ee46d3b9e7b7a : SUCCESS in 21m 45s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b99ec8cf4bc4622bbc04c77a4b9112b : FAILURE in 38m 38s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/de2b2d5049fc4d959360ddd854e943df : FAILURE in 11m 17s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b9c0ff1f6a40ed9386fc042b7ddafd : SUCCESS in 31m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43916709363b439a824a790623ad0f49 : SUCCESS in 31m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64efb6e9d6b048299a6f0983ca8fb1e4 : SUCCESS in 5m 53s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e8cafa9e4034fe9b835b7a803ffca3a : SUCCESS in 1h 52m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"67878b2397df5eafdb943bddb1d613f2d89fc3b3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 09:03:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"930f6d6c17aeec67be9c3d4b0480d393579c59af","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 10:09:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5ecfe2f996462135f6c0d153408a5ac3cf5777e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:58:51.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd97698ccbbc40618c7b668374a18257 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41730760ac9f4118851d1afe969aa9f0 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1f816ac4373420ba85d00bedee78061 : SUCCESS in 4m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/59d6f96bde0441148a3748ce3dcb7c31 : SUCCESS in 16m 11s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ab181b61eb4741d3bd2a0d8e0c8c17d5 : SUCCESS in 12m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ff58c3827be4a2b9a4e92e814770938 : SUCCESS in 11m 04s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e26d89eb5755469895d099cdc2e43484 : SUCCESS in 11m 40s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c3929d65c024e16a62b3ef442dd9e76 : SUCCESS in 10m 25s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33b168f163e84fb0aa18e0849e745a0a : FAILURE in 30m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e9e1e4781c7f448f9fa5040bbea65dee : SUCCESS in 1h 06m 56s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/da8939fe89e3465780c84a032c3bb20e : FAILURE in 46m 42s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a903b11751d943f9b23921753031dd48 : SUCCESS in 19m 30s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/eacc14975ef847d3b1fba64eb11ce166 : FAILURE in 36m 08s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/12e2a3e4cafa4f4d93913b7c49f54930 : FAILURE in 10m 05s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05d997c8ff7b4347922e3d185ebffe4d : SUCCESS in 24m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/275cc68e84fc4c81908069b063149960 : SUCCESS in 19m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d555cd7130a94de58edc4f06f0737768 : SUCCESS in 4m 33s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5340be96ffe54d13bcf5b81eb40076b3 : SUCCESS in 1h 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"dda1e9eaebd44ee2d428fad8c91daae917d6a409","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-04 04:37:50.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f4138f6ab8dff32f9815d0660989e3e90c98240e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 06:20:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b60f82ba685547a88b23859e6bed8e23 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb0138015eae483abdc28194e48fec39 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de273e388b934e6498fb5bdbf5c4ca42 : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b3574802907421cab415a16d70c95db : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d7c76ac813a4727bdae0780820d9027 : SUCCESS in 15m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa315f0fd29c46c3bd83a3e5f46de0b5 : SUCCESS in 12m 33s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8410daceb6940b7a2fd49a2fc06b257 : SUCCESS in 12m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7ede9100b674fdd822b4dfe6d52c3e9 : SUCCESS in 10m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26c83ced29d14470804da8040f48bb5d : FAILURE in 32m 25s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8f9288d63f74e88a154c10a16d00e54 : SUCCESS in 1h 12m 54s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5633a543af314cfc94c92d867784ef64 : SUCCESS in 51m 05s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/453e634242b9446c91d6d3c16d14b7aa : SUCCESS in 23m 59s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8d32d4086bb4e7fa2f153880b6b64e9 : SUCCESS in 30m 51s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a169a96c3e6403a91137a3d7393a13f : FAILURE in 9m 11s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c76872b2c0242b99189c4e2bdd876fb : SUCCESS in 24m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f8e139e08b5407bb1240b533a1dfde0 : SUCCESS in 22m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2009ccc769974c42964014406193b0b4 : SUCCESS in 5m 11s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/17c2bb4118be440b8b04a642c5615c83 : SUCCESS in 1h 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b10eecd104ccd7ef00caa269bcb5da572c8fed60","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:40:08.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"1009e1f4a54227e1ec2e431c048f0802c329bcac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:41:33.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4325633a190bc4cbbb5064ebb0e33371c0608f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:52:36.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/624e9415cee34807a64ff32de3301864 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63185e1e91d549238c6453f35cf68d76 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee0f699620c64c15ad3b9939794a58a6 : SUCCESS in 5m 53s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/15ad46ad8888414780e657f8d7fb0425 : SUCCESS in 13m 58s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/58a6146e7d0e43319331394954b1f983 : SUCCESS in 13m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db602e6b459d4156aa575be1c4ac5f32 : SUCCESS in 11m 20s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3986641435c4a6ab1aeb97496525135 : SUCCESS in 10m 55s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2a08acdd6c44ffdb93aea50b371814a : SUCCESS in 1h 05m 16s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/db7b205464f24cb794bbde79e8da8dc6 : SUCCESS in 31m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5683dba15d84f8ab43bad305af4c014 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":11},{"id":"ec512519750d181d300844fb43c4786885cdf0db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:52:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"adfa9817ea859909abe62e14973b9395718fd2fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 00:53:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/156d67fe0ad84abd996ee5799ed23c5c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"2b9f2a0813ea72605f8d673b5dd88defc4146262","revisions":{"072c8c0cfbe02a248698d7b76d4354685b199d20":{"kind":"REWORK","_number":1,"created":"2021-11-03 07:15:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/816434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/1"}}},"commit":{"parents":[{"commit":"612142e30ef1568dd5c75970a5636c2a4872a8db","subject":"Bump puppetlabs-inifile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/612142e30ef1568dd5c75970a5636c2a4872a8db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:46:36.000000000","tz":60},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/072c8c0cfbe02a248698d7b76d4354685b199d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/072c8c0cfbe02a248698d7b76d4354685b199d20"}]},"branch":"refs/heads/master"},"8e5fe2de31221b63577c67d7d09331c11a7508a8":{"kind":"REWORK","_number":2,"created":"2021-12-06 14:50:49.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/816434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/2"}}},"commit":{"parents":[{"commit":"612142e30ef1568dd5c75970a5636c2a4872a8db","subject":"Bump puppetlabs-inifile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/612142e30ef1568dd5c75970a5636c2a4872a8db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-06 14:47:01.000000000","tz":60},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nTODO: additional db is not created in postgresql\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8e5fe2de31221b63577c67d7d09331c11a7508a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8e5fe2de31221b63577c67d7d09331c11a7508a8"}]},"branch":"refs/heads/master"},"e5a03b9f9368f83fa18597de04801377e88e3ff5":{"kind":"REWORK","_number":3,"created":"2021-12-06 17:02:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/816434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/3"}}},"commit":{"parents":[{"commit":"612142e30ef1568dd5c75970a5636c2a4872a8db","subject":"Bump puppetlabs-inifile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/612142e30ef1568dd5c75970a5636c2a4872a8db"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-06 17:02:00.000000000","tz":60},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nTODO: additional db is not created in postgresql\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a03b9f9368f83fa18597de04801377e88e3ff5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e5a03b9f9368f83fa18597de04801377e88e3ff5"}]},"branch":"refs/heads/master"},"9da53e0fbeda3c69461acbd68395ecde5cf5eb43":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-12-06 17:07:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/816434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/4"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-12-06 17:07:31.000000000","tz":60},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nTODO: additional db is not created in postgresql\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9da53e0fbeda3c69461acbd68395ecde5cf5eb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9da53e0fbeda3c69461acbd68395ecde5cf5eb43"}]},"branch":"refs/heads/master"},"f2f4809ebbcfcad4d88600c692519eeb16f90200":{"kind":"REWORK","_number":5,"created":"2021-12-07 01:55:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/5"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 01:55:03.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f2f4809ebbcfcad4d88600c692519eeb16f90200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f2f4809ebbcfcad4d88600c692519eeb16f90200"}]},"branch":"refs/heads/master"},"b6c598294095466e79af6c6d0a51c23db1a5d818":{"kind":"REWORK","_number":6,"created":"2021-12-07 02:40:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/6"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 02:40:26.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b6c598294095466e79af6c6d0a51c23db1a5d818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b6c598294095466e79af6c6d0a51c23db1a5d818"}]},"branch":"refs/heads/master"},"c3140ccb618430235bf81e3a897b5a7618cf89c1":{"kind":"REWORK","_number":7,"created":"2021-12-07 02:57:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/7"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 02:56:58.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3140ccb618430235bf81e3a897b5a7618cf89c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3140ccb618430235bf81e3a897b5a7618cf89c1"}]},"branch":"refs/heads/master"},"c3a4bb2624b2328755f8167ad667ab18f7a219d6":{"kind":"REWORK","_number":8,"created":"2021-12-07 04:08:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/8"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 04:08:01.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3a4bb2624b2328755f8167ad667ab18f7a219d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3a4bb2624b2328755f8167ad667ab18f7a219d6"}]},"branch":"refs/heads/master"},"ff79aa4d70959334879a588d2841aba6a20d1f85":{"kind":"REWORK","_number":9,"created":"2022-02-28 14:40:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/9"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-28 14:40:05.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff79aa4d70959334879a588d2841aba6a20d1f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ff79aa4d70959334879a588d2841aba6a20d1f85"}]},"branch":"refs/heads/master"},"10e66c19dea947dc0a67b1f2b81ff9af97cc6e99":{"kind":"REWORK","_number":10,"created":"2022-02-28 14:44:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/10"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-28 14:44:16.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/10e66c19dea947dc0a67b1f2b81ff9af97cc6e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/10e66c19dea947dc0a67b1f2b81ff9af97cc6e99"}]},"branch":"refs/heads/master"},"2b9f2a0813ea72605f8d673b5dd88defc4146262":{"kind":"REWORK","_number":11,"created":"2022-02-28 23:04:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/816434/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/34/816434/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/34/816434/11"}}},"commit":{"parents":[{"commit":"3513ff3dabad3423ab26fe17b8f6265bbc72a31d","subject":"Fix dependencies related to openstacklib::policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3513ff3dabad3423ab26fe17b8f6265bbc72a31d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-11-03 06:17:08.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-28 23:04:09.000000000","tz":540},"subject":"Add parameters to configure amphorav2 driver","message":"Add parameters to configure amphorav2 driver\n\namphorav2 driver was introduced in stable/ussuri in Octavia, but some\nparameters to configure the backend server to enable persistence were\nstill missing in puppet-octavia.\n\nChange-Id: I8a9bf0b2c17cd35bc09fbb0270c4ef4ed4e491bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b9f2a0813ea72605f8d673b5dd88defc4146262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b9f2a0813ea72605f8d673b5dd88defc4146262"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
