)]}'
{"id":"openstack%2Fpuppet-octavia~819651","triplet_id":"openstack%2Fpuppet-octavia~master~Ic75b471a4853b5176804ff6163c24285ce2703ba","project":"openstack/puppet-octavia","branch":"master","topic":"networking","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:06:29.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ic75b471a4853b5176804ff6163c24285ce2703ba","subject":"Support more [networking] parameters","status":"MERGED","created":"2021-11-29 14:58:35.000000000","updated":"2021-12-08 02:55:12.000000000","submitted":"2021-12-08 02:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819651-networking","meta_rev_id":"c2f5e4844e395f39c59fa3ed047b81ea4c4d2949","_number":819651,"virtual_id_number":819651,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 02:51:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-30 09:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:59:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:59:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 19:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 09:03:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:59:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5200363c5143ffdc5bb2862ae9aac68a0131149e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 14:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d36e8ed2b2054ae595f401876bd33ac0994b0bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 15:37:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cae058040cea6d9bfce1f2da1ba6391a9e2c9cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:03:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7376e06a4af54a5a8abaa45efaba0ffc : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24b577c6782946eda3dea85c8d266b0c : SUCCESS in 3m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47ca3e6493404c9693ff29779fe61b81 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e0de07763ea423abf85ebe91067ef56 : FAILURE in 13m 31s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a4f9ce63de443db9dbcd524dec7b972 : FAILURE in 12m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/558007fc7d0c45568c931d6681d3781a : FAILURE in 9m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc7624b056443fda30f255ab741916a : FAILURE in 8m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39fb7b738c9b4670ae03fbfde40aba29 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a221285e121c4ac1befac06b9fc0699e : FAILURE in 15m 08s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e388ca1549dc4434b342a5dfed6631b4 : SUCCESS in 32m 17s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c011d4fb72c6434f8868a6a4045589ad : FAILURE in 15m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/994c497808ad41f5923cb1d5774f55a5 : FAILURE in 14m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2b66b0374d5426b951dc7fa7502653d : SUCCESS in 3m 27s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d62a2c1a1b70479bab9614cdaa63b566 : TIMED_OUT in 3h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ce5309fe839c4ba98463eacf4e9a577be91a97f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 23:02:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5e40c10093eff79a1d3ce9215aaafc1dc6ab5bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 00:23:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac7c6056f9697b9c556a49d1b6b84847d0d0ffde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 03:43:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc9823726cdb488dadcdbde1a550d948 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509e95bc3a994a2090066beeb3c06698 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91df34dd521e4711af170462458e016f : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc9affabf1ce45a7ac8cb2996e6c05a3 : FAILURE in 10m 26s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/da4f6e4de40a4c9e8fa87069bd3df59a : FAILURE in 16m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557a210e6fa24754822b47dabe1a85b0 : FAILURE in 7m 58s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9a84798e48244c193d693fc0509daa3 : FAILURE in 10m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b03a01ea619c4d1c9c68bcf126d79afd : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c795bf0a81c04ad08ff0574a336ab627 : FAILURE in 12m 24s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b9e3177e0c84a3a94658f80c604a672 : SUCCESS in 23m 24s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3b3b98efe46490aac156fb715b2f405 : FAILURE in 21m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f05301bc7bb4e4c80d060d75de1b6fd : FAILURE in 11m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c980177e8744122be2f56dc1042006e : SUCCESS in 4m 01s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ed9a2ace14434793ba87563944b8c6c4 : TIMED_OUT in 3h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ae01f6d1232cfa72e1f14c3077b8941ccfd672a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 04:06:21.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d34a090c09ee80ebf9f75e70813e97cf801fb02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 04:08:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6612fa41479508a8c69e62a74e6a4378975d4a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:31:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dec86433ee464966af13cb443380560a : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace4bcae9f1f4e3c80cb2c0b7bdbaccf : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a3d63fc9fe04012a9971c491c605be1 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/87b38e046dcf47809821c4872dcb90ec : SUCCESS in 12m 13s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/89a52d8c8a6e472b99540019d105f0e4 : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f52adae672241198966326824a31e7a : SUCCESS in 9m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dda86208cc44b3487391e009ffa0bf9 : SUCCESS in 10m 22s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecae42e0308741f4a6a6b078a022ff18 : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00a0a8674ba84870800eaebf0a98e4ac : FAILURE in 13m 59s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0900492faa140479eafbac8c664a452 : SUCCESS in 29m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05c902e1fc2746228810761e9298a2b1 : FAILURE in 13m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/214ca44281a140b78023ad1c9966ab08 : FAILURE in 15m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0375a9badb924f10930a9690be253d0d : SUCCESS in 3m 29s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9f11860451f246b798509126d77912ac : FAILURE in 2h 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"32321994e38b4e89d4fd3856c6685665c34cce3d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-30 09:03:32.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"910d5d5bee00995e1ae026b9b3d38d5abb6ecccc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:59:20.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f45d2475f275e0a157d8dbb549790dabbb69d5c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:06:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a5431c30489ff675324b701afa1e9dfa0cb3aaf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:05:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41539ed546be46798ab69bfa3b02d4d6 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46d255870b2949c8a3b00de88298baf2 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a499979e45e438b82cc9dc9f5f624d8 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/74ca39c07bb64d45a2011242430bfd76 : SUCCESS in 14m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4daa3f28aa814f4f81a2caf6461c9f79 : SUCCESS in 15m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36fd2a341dd645918bfeb40e052a0d26 : SUCCESS in 10m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d52cde40a83246b8974f55a6afbc44fd : SUCCESS in 9m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30d5bb11cd5f4a53b4bd0c6babe89638 : SUCCESS in 10m 26s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20b77cf96ce34b139c99057a7459b094 : FAILURE in 13m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/049b4f962ce34ff39c3bef62f99e6274 : SUCCESS in 27m 36s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8e9906d2bfe4e58a42e7727e41cbfd8 : FAILURE in 7m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b2190dc55f74bafb81776921fd48ea3 : FAILURE in 18m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06123253c0d487da0a81faa77ca7078 : SUCCESS in 5m 15s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/60099800da474026a17a188702c10fe6 : FAILURE in 2h 52m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f9fcc641bb462c55c15372d238b8889fc1f3f3ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:17:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2e0a17e58ba72c66ad961d61c69b34d0cf9eeb53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:51:35.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff41aed5c88241429409819c415f536c : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c2c7e6ca5c24740be3dafbecc5297a1 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54ea7d48b4f547de8de1672deab1c098 : SUCCESS in 5m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9bcb7691088346988db8e43645040781 : SUCCESS in 17m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/80717e6f84744ac49017b21a5ebf25f8 : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/665cf0e3f16b4a479cc1c465555834f7 : SUCCESS in 13m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/290cfba0a75744e6b6ce356999e77162 : SUCCESS in 13m 47s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/98c8abcb97a3498a8a4e467c7e1fb7ef : SUCCESS in 32m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ae846ad6f154afca4261ace47421dd0 : FAILURE in 18m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb633dfcadbd44228aac591b23c9825a : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"d3aabd3da813b264e05772b5bbf44d4fd078c330","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:51:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c2f5e4844e395f39c59fa3ed047b81ea4c4d2949","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:55:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef7914102465453dbf4bb1ae8a748f29 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4516e0cf9cbf22c78b07e7df6dda2f46afbd99ff","revisions":{"50276568ec15378efca23cfc0462ecb3b39a3a46":{"kind":"REWORK","_number":1,"created":"2021-11-29 14:58:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/1"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/50276568ec15378efca23cfc0462ecb3b39a3a46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/50276568ec15378efca23cfc0462ecb3b39a3a46"}]},"branch":"refs/heads/master"},"4392380d65d20ee6a22b000e10115a7892c1a697":{"kind":"REWORK","_number":2,"created":"2021-11-29 15:37:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/2"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:37:15.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4392380d65d20ee6a22b000e10115a7892c1a697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4392380d65d20ee6a22b000e10115a7892c1a697"}]},"branch":"refs/heads/master"},"b26d2555aee48224b26b4f8ac9483d550fd26037":{"kind":"REWORK","_number":3,"created":"2021-11-29 23:02:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/3"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 23:02:48.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b26d2555aee48224b26b4f8ac9483d550fd26037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b26d2555aee48224b26b4f8ac9483d550fd26037"}]},"branch":"refs/heads/master"},"f242f0eb29498a502eb226b9fd8e92591adfb1f6":{"kind":"REWORK","_number":4,"created":"2021-11-30 00:23:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/4"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 00:23:14.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f242f0eb29498a502eb226b9fd8e92591adfb1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f242f0eb29498a502eb226b9fd8e92591adfb1f6"}]},"branch":"refs/heads/master"},"772002e78350e915ab4139a1405dfffa4988fed2":{"kind":"REWORK","_number":5,"created":"2021-11-30 04:06:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/5"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 04:06:07.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/772002e78350e915ab4139a1405dfffa4988fed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/772002e78350e915ab4139a1405dfffa4988fed2"}]},"branch":"refs/heads/master"},"4516e0cf9cbf22c78b07e7df6dda2f46afbd99ff":{"kind":"REWORK","_number":6,"created":"2021-11-30 04:08:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819651/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/819651/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/819651/6"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 14:57:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 04:08:02.000000000","tz":540},"subject":"Support more [networking] parameters","message":"Support more [networking] parameters\n\nThis change introduces the new octavia::networking class to manage\nparameters in the [networking] section.\n\nChange-Id: Ic75b471a4853b5176804ff6163c24285ce2703ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4516e0cf9cbf22c78b07e7df6dda2f46afbd99ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/4516e0cf9cbf22c78b07e7df6dda2f46afbd99ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
