)]}'
{"id":"openstack%2Fpuppet-octavia~819653","triplet_id":"openstack%2Fpuppet-octavia~master~I139ed46b7acbb81d9c01c8708f91af9e5d2900a3","project":"openstack/puppet-octavia","branch":"master","topic":"compute","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-07 23:06:24.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I139ed46b7acbb81d9c01c8708f91af9e5d2900a3","subject":"Add support for [compute] parameters","status":"MERGED","created":"2021-11-29 15:01:36.000000000","updated":"2021-12-08 03:11:16.000000000","submitted":"2021-12-08 03:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819653-compute","meta_rev_id":"5da847534812cc8f98cbb10cda9585b82a50d10b","_number":819653,"virtual_id_number":819653,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-08 03:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-30 09:00:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-12-07 15:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-12-07 15:59:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 19:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 09:00:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-12-07 15:59:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e173c35d758e6b65f33ee4f12d14720d490463b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 15:01:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2744ceb963df2e14f629278ffceeb5580153f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 15:37:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0154caedcf8cc8ab18f1fe115406d01f5220057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 19:03:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00cc7e14b7ae479ab0f96a32d8193404 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59c1b29e5d8c4e3487b1b8e867075bd1 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a945c773506a48cf9b6e43edefa619a1 : SUCCESS in 3m 45s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/54ac600b0fd04d229e697063caf206d4 : SUCCESS in 12m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1f6c95ff9a84182a0f11cd08e93fca3 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d106b04e0f7141289673b0d721c31b1c : SUCCESS in 8m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/137a137d192f4bab83bb70f6d6c57d83 : SUCCESS in 8m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40a81f4c57f74372a6b63636ddb9aa06 : SUCCESS in 8m 35s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6d0fcd480f34d74986e4b4f6b085d9a : FAILURE in 13m 52s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3edf63774d1c49b4bf1df385d1cb5453 : SUCCESS in 27m 26s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57238b43cb1f45b6b1aca8216fef239b : FAILURE in 11m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921501150e9c46a18f69f447c18f69a6 : FAILURE in 13m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffcaf32ad36c47f7aee177a77c987ca5 : SUCCESS in 3m 34s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d02137db987943a09cce542be2a0d20f : TIMED_OUT in 3h 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77d8216afa8359fbfd904417f9ef6a448363cf28","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-30 09:00:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e9be24c72bc0f88c045d06d995ee1eda81ac15c3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-07 15:59:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0973ebf36b550fb9ea848631fe9349a70d215807","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 23:06:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a099dcb957773420c24e4d888ab2d5a461e31343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:25:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faf5237862224fe0aa8cd492478a7384 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e0f05c913be48d8a239d662da54a616 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ac670763554d36b377e06cf68658e4 : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc9bef372b074e43adbb52fd4bf2b870 : SUCCESS in 12m 41s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae97b243454f47e59b0263385c952226 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c76dfcdff2c54583abde8566cfce7cb4 : SUCCESS in 10m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b604fa6607d40e491230e55cb876d6b : SUCCESS in 10m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02bd2afb406b4d9788660a06bece8cab : SUCCESS in 12m 48s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/810f91f58a874c869c0538de552763f2 : FAILURE in 15m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cedb88310d04c94878e33f588a2cc42 : SUCCESS in 28m 47s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4740fe13fd924e1c9abd70b89776bda8 : FAILURE in 14m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/922b93a9a821457892965c062e402b44 : FAILURE in 14m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7c6130bba4485ebaa3e6dfa722c8a5 : SUCCESS in 5m 12s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21c9abe671674bf1b45e193f01b0c6e6 : FAILURE in 3h 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f85d2cd9ead5b011b12742c1d3a58885dba5c9eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 02:28:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e1519eed69813a77ca64b2084594981893daf765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:06:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c477ca798ea84838a0e5c1430a21b853 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94b9b11641a440878c55240215082582 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb45f05dda7f404a8db7ff496ee6dc55 : SUCCESS in 5m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e895a65cf854df5b566a4c72bbafc06 : SUCCESS in 13m 33s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b0002fb8b5b4fc78e4487424fc26cb6 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bc48691252b40239926c5a9b3a2c5a2 : SUCCESS in 12m 25s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be2cce22c87f4682bc453621ba7c0ad5 : SUCCESS in 14m 06s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea170dd3fe024b44802d8e4285d315e3 : SUCCESS in 31m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/441faf93aae2499cb36b31360c351199 : FAILURE in 14m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/908b70ab04384e0da8ac5e2138ff5327 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"bc9e5642dd2a72b19da4c434d0709360910551c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:06:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5da847534812cc8f98cbb10cda9585b82a50d10b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:11:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c78787809c43f498e2f457e7e308aa : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cf24d758afa627ef455cd0dc178053229d25471b","revisions":{"9e967b390b5b1d1a8cc9571af63f8390105644da":{"kind":"REWORK","_number":1,"created":"2021-11-29 15:01:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/819653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/53/819653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/1"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:00:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:00:57.000000000","tz":540},"subject":"Add support for [compute] parameters","message":"Add support for [compute] parameters\n\nThis change introduces the new octavia::compute class, to manage\nparameters in the [compute] section.\n\nChange-Id: I139ed46b7acbb81d9c01c8708f91af9e5d2900a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e967b390b5b1d1a8cc9571af63f8390105644da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/9e967b390b5b1d1a8cc9571af63f8390105644da"}]},"branch":"refs/heads/master"},"cf24d758afa627ef455cd0dc178053229d25471b":{"kind":"REWORK","_number":2,"created":"2021-11-29 15:37:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/819653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/53/819653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/53/819653/2"}}},"commit":{"parents":[{"commit":"2fb4a5b1e07478a1e67d4897f26c603343dfa937","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2fb4a5b1e07478a1e67d4897f26c603343dfa937"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:00:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 15:37:40.000000000","tz":540},"subject":"Add support for [compute] parameters","message":"Add support for [compute] parameters\n\nThis change introduces the new octavia::compute class, to manage\nparameters in the [compute] section.\n\nChange-Id: I139ed46b7acbb81d9c01c8708f91af9e5d2900a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cf24d758afa627ef455cd0dc178053229d25471b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cf24d758afa627ef455cd0dc178053229d25471b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
