)]}'
{"id":"openstack%2Fpuppet-octavia~820620","triplet_id":"openstack%2Fpuppet-octavia~stable%2Ftrain~I18e2f3376fefea600a1ad5a931ce52310b44f51f","project":"openstack/puppet-octavia","branch":"stable/train","topic":"at-interface-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"32114":{"account":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"last_update":"2022-03-21 13:52:58.000000000","reason":"Lewis Denny replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-28 22:46:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18e2f3376fefea600a1ad5a931ce52310b44f51f","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2021-12-07 14:25:10.000000000","updated":"2022-03-28 22:49:37.000000000","submitted":"2022-03-28 22:46:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820620-at-interface-stable/ussuri-stable/train","cherry_pick_of_change":765271,"cherry_pick_of_patch_set":2,"meta_rev_id":"e108a7d58de63ea102b52113d1646db615ab857d","_number":820620,"virtual_id_number":820620,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 22:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-21 19:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-28 20:24:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-28 20:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 14:25:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"state":"REVIEWER"},{"updated":"2021-12-07 17:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 19:08:57.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-28 20:24:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"69a55424cbcd712390a64f48dfdfd4ca4fb20b0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 14:25:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"b83d80ef35a197c0d9e51bd7567469a6ab5659f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 17:18:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b770328411fb4dc5965f588150a12e1c : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7592bbb047c44b498316ec53b6098f6 : SUCCESS in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123c895135af4e8a940554d9675ae32d : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9e2afa56f0d84fafb721cb08892b1bef : FAILURE in 16m 17s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f92f5d86597347fd8f5c2622e962245b : FAILURE in 14m 26s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ed314af93d641499574586eac868083 : FAILURE in 14m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaa69f3d107a4e76bea757a8e5732450 : FAILURE in 11m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03e9e82bd67643519ea4c80e41803659 : FAILURE in 13m 25s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/946a61e6a764453abb94619051c65372 : SUCCESS in 24m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95d1a1dfc0fd49e39fa804a65ac347ba : FAILURE in 9m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20521557116646bb8077ed2cb6ad04ac : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a34904a62fe9671b9f8ef0cf451992ab34c6d1d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-11 15:44:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c52f8e40b4f2aeb294a4310b4c749ec9160bbecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 17:11:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8f44af0f5f2407d93cf8fc5c4141625 : SUCCESS in 7m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af09115250d84d8a962cb1f5d76efd1e : SUCCESS in 6m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74925de69f534046906b1e1038706b9d : SUCCESS in 6m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a6ee65896b9b4589b8754091c3268399 : TIMED_OUT in 1h 05m 31s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fbb39db0ca64b3d8e92dee20e695bf6 : FAILURE in 16m 22s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/599d4894d30e466ead8ff8fb02b78840 : FAILURE in 17m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6ed676a0a5e45029715399a8926535d : FAILURE in 38m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0ac4fa3118a495881bf4cff0a7ecef2 : FAILURE in 41m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a55420b5716c492b8b8cb34bf8538159 : SUCCESS in 26m 16s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b61f85506f14c88a691dff242dda57d : FAILURE in 13m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5bedf7313dd4ebcaed90f01e242df53 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8e2aaa05a93f39f44e6db29c130e1e9ca51239c6","author":{"_account_id":32114,"name":"Lewis Denny","email":"ldenny@redhat.com","username":"ldenny"},"date":"2022-03-21 13:52:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc9556ad4318d10705feb173e464751a7731e78e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 14:24:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aaf0d8a330e441fba6951abb44d3dbf9 : SUCCESS in 6m 24s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a333c29d4854dd1bef5306393cb1d9d : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df81156bbbb84bd6a727605f2325b95e : SUCCESS in 6m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a0b3bb98d4a242a4bc6a09306395ff9f : FAILURE in 16m 19s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/38fa6df9175240b1ba98cd605bb7519b : FAILURE in 13m 58s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/1644db72d371464ab2ce219ffb1d5c7e : FAILURE in 13m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/487ebb7d29e04a15a33bf85bef6ef282 : FAILURE in 16m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3837d56901144d4086ead1f13c20693a : FAILURE in 11m 52s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9e431b868b02473092a7f436316f2458 : SUCCESS in 27m 01s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c73f619f886f4ab29bf2bfbca4d15bd3 : FAILURE in 8m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83e0ed7c00c546d3a6a79e2080b79ad7 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"94505a71ffeea6c0b57459ce4298f5d18d2d7a76","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-21 14:25:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57b43496138fd4c86acf416961103b567d3ed29a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-21 14:32:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1142bcaa83e7186869ffb1d6578b9195a9741284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 15:00:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7040077a26a2401186c2bbf7503ba8d1 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d26c7fd1de1d491c9eaeba9c77fc3587 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01dae612b96f46ecba8673723e59d5be : SUCCESS in 5m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/18e910f42e7f40cc8cef689849e6cce0 : SUCCESS in 21m 30s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/72b0920d08a94683a245604b4f9225e4 : SUCCESS in 16m 20s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3f4abf423ef4726a20da30649e3f692 : SUCCESS in 15m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d8b38a2edc64435bfa4c5036caf6b15 : SUCCESS in 13m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b6f7e3597534db9a8b722e9b6d21c71 : SUCCESS in 11m 46s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e024d25ade104cdebe497aa03217f4f3 : SUCCESS in 22m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c92e49f694fb43fcb01e1675bcf2a7e5 : FAILURE in 8m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eb9d05380514aeca3990c6b48b12b3e : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"ab9630c23eabe1b385eea1848f9ab344537ea9b4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-21 19:08:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1cc994f0a83e9391e4f23172d55c33e7e52e11c5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-28 20:24:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8c31348bd913c1dffa7244d505d8ed0a860c555c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:12:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6290984c257d61362428a42779e8df2837f3c767","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:46:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a2c71b2dea4417097a79a7b5741376e : SUCCESS in 6m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e85f9c67f9e4439819d6eff4967f1cf : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80ca4b2892ce408c9724e885f547e04c : SUCCESS in 6m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/67456b77ad4541db941b41cb6301619b : SUCCESS in 18m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8966cb33454b4cc893d96e26da47ed2c : SUCCESS in 11m 49s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bcb4b1ef6a7c47cc83357cad38714de4 : SUCCESS in 27m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5788341f947a4e0d84b5b8ea72aa3cd8 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3a3eba51e0d85366434710358cfc1f90fcc436ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:46:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e108a7d58de63ea102b52113d1646db615ab857d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:49:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a27cc6fd99324cd0b8dc08744c7f8ebc : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db40598a78a4ff4e7293e9084c13719f89c22f0c","revisions":{"fbe1c592be985b549e5de5f5e945bec332ecdbae":{"kind":"REWORK","_number":1,"created":"2021-12-07 14:25:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/820620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/20/820620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/1"}}},"commit":{"parents":[{"commit":"d6e88fb5399aadae2dceb8e219a239a825609397","subject":"Add support for [oslo_middleware] enable_proxy_headers_parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d6e88fb5399aadae2dceb8e219a239a825609397"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 11:27:21.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 14:25:10.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I18e2f3376fefea600a1ad5a931ce52310b44f51f\n(cherry picked from commit e21746f37588caffe9c0cb4c4a02e351f65a348f)\n(cherry picked from commit bb73a687352ab7514e8defb80a31967f50c4db8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/fbe1c592be985b549e5de5f5e945bec332ecdbae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/fbe1c592be985b549e5de5f5e945bec332ecdbae"}]},"branch":"refs/heads/stable/train"},"db40598a78a4ff4e7293e9084c13719f89c22f0c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-21 14:32:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/820620/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/20/820620/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/20/820620/2"}}},"commit":{"parents":[{"commit":"276185bb30f6bbc76f5fcf413bdeba691508c068","subject":"Do not expect database/idle_timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/276185bb30f6bbc76f5fcf413bdeba691508c068"}]}],"author":{"name":"Lewis Denny","email":"ldenny@redhat.com","date":"2020-07-14 11:27:21.000000000","tz":600},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-03-21 14:32:19.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I18e2f3376fefea600a1ad5a931ce52310b44f51f\n(cherry picked from commit e21746f37588caffe9c0cb4c4a02e351f65a348f)\n(cherry picked from commit bb73a687352ab7514e8defb80a31967f50c4db8c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db40598a78a4ff4e7293e9084c13719f89c22f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db40598a78a4ff4e7293e9084c13719f89c22f0c"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
