)]}'
{"id":"openstack%2Fpuppet-octavia~823938","triplet_id":"openstack%2Fpuppet-octavia~master~I6a2686a4e33c75d863ae146a3af37d416e01a655","project":"openstack/puppet-octavia","branch":"master","topic":"connection-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-24 08:45:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a2686a4e33c75d863ae146a3af37d416e01a655","subject":"Add support for more connection tunables in [haprpoxy_amphora]","status":"MERGED","created":"2022-01-10 07:59:18.000000000","updated":"2022-01-24 08:47:08.000000000","submitted":"2022-01-24 08:45:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823938-connection-opts","meta_rev_id":"89b5a8b95ad02d8902e220aa41148922cf8926de","_number":823938,"virtual_id_number":823938,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-24 08:45:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-01-21 21:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-01-24 08:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-01-24 08:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-10 10:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-21 21:32:21.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-24 08:12:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"251d1c82fa4bb86cbabd85146363c280fed3cd77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 07:59:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f217d1cc444ec5875a49abb9a86e523621bd367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-10 08:00:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b36498abe3e6e390b3265325e5e0069e38ec390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:42:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/262522c592674dc7993389a4cbaed00d : SUCCESS in 3m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/010183a8cb86435e806e5dad43de269d : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb0fcb3262684915a559086aafa9c635 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0af2eb278c73445b833f7a1f702a22e0 : FAILURE in 12m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b25e4e71446421ab6a6c63058832ca7 : FAILURE in 11m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77aba2565af14c2aac8296f067ffb146 : FAILURE in 9m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e827f4ef89e41a1ba79a315ea6849df : FAILURE in 8m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8192e358b003422190d1e0305fc511fa : FAILURE in 8m 54s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb73ea7039794e11b226a7fbfd32f162 : FAILURE in 15m 31s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b9b7152da61403084450f5095e9b77f : SUCCESS in 26m 08s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9745df35ce134e9d9fa6a20414c4cdf7 : FAILURE in 16m 19s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129c64086eaa4a01a18b33b68045af3c : FAILURE in 15m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd55e510d884d678fed3e630f21af72 : SUCCESS in 3m 43s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/701472536ffd46c1acf5753c010fe6d4 : SUCCESS in 2h 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"96b42fe137f3d9370ccf4c7eb9cbe45a47b027ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-16 02:05:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cd9aa1d9f535590da1ef4e1c20d41b942d58fd58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-16 05:16:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb9319662f24f9bad08734fd3d1251d : SUCCESS in 5m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bacdbcee91e47db992722f840694b46 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96142987afc0477194a408c7cb730316 : SUCCESS in 4m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c57927bcb69f4606b04a0d98aab70d93 : SUCCESS in 13m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f239b99a5c74283a2cc81a62011caea : SUCCESS in 13m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/347ad59a0bc9492e9a09d92e0015514b : SUCCESS in 12m 12s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa4ba7862b1c4e9e9c949535d62f794d : SUCCESS in 10m 21s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aafb04c6504041988329ccb54aebbc23 : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29fc660247a440ff8ccab52c42f1539e : FAILURE in 15m 12s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e086ca3375714860ae463b7a1d9b1bab : FAILURE in 41m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18bba982b5e64000a993da1339f78456 : FAILURE in 16m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6360f566a2b146cabc192a4a7a1a1bcd : FAILURE in 20m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b384b85e85e40e3b69144cb17bc07a6 : SUCCESS in 4m 57s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b3b5538b6f149be96227fe97a6fbdb4 : SUCCESS in 2h 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b24bd9ea891e4a6a0f1ec5ad787c6f99585dcec","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-21 12:13:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b0b86bfde8e6f5041ff07cb282262dfc4181c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 14:22:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09a4d5ee89614cf59bfba60a107559b3 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b04455222b5412593078d50be87a047 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb4b484e3fb7435d91d3530ec5d14c30 : SUCCESS in 4m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1f10c36a5dc46f8b501a3fae705b353 : SUCCESS in 14m 47s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b3235314dc0485d89485f1a5dd4058c : SUCCESS in 14m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f0226d34eca4066b5fd2f8d6296663a : SUCCESS in 10m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ad57ffe26240448f4efa5418f7bee9 : SUCCESS in 10m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca21d394aa4b454f88a95ceafc81e044 : SUCCESS in 9m 57s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0767650032254e36bcb31550ccb9f9b8 : FAILURE in 15m 24s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4662da5bd97948fb942cb86f602c70d2 : SUCCESS in 27m 03s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93445981cf454a809205b5fd39a6427c : FAILURE in 17m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37b1f5d1825641cbb755142d02102037 : FAILURE in 16m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/898aec5310ca43658add91baeb624d31 : SUCCESS in 4m 21s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a153ea7489ca42568b21c15be5d6465b : SUCCESS in 2h 05m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce39a7f03048b2b4276e62bab7d9eac7bea94b41","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-21 21:32:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"88fc09ec7a2c6667f771eb5b2e8aedad39ad1846","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-01-24 08:12:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"022092959043948156ca1be18529ecf0d409139d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:13:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ca2fbd9bebdd83bf0c0bb7882b96fc81fc0fb6b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:45:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df9db8cd133444ca94695638de7c3ad1 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78377a396c7e4bb49c79d1ba13985a69 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5aa0b1d70a9478e91c0899d4f06fd4a : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f61549198ac4f91899ca3f55bd0470e : SUCCESS in 10m 48s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/769f6ddd64534609977175ea9460d17d : SUCCESS in 10m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a308810d049941d7ac0264a358d9bf31 : SUCCESS in 12m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c07e26e22c8942d297de960f4da34a4d : SUCCESS in 8m 05s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/49c3664dd9534d2b97900eababbcbd9c : SUCCESS in 28m 32s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a4f478e57a4b858a6108e464ee1d9a : FAILURE in 13m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aeb43eb5342249ffa9bca69d86512112 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"97f123ad3f1d36d4fef0a87f794d9de738c534b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:45:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"89b5a8b95ad02d8902e220aa41148922cf8926de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 08:47:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01f69d859e4046e68f81af832b55eaac : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ca3c4cc1c166c473ecf3935936c3ef4946d4c43f","revisions":{"db2102f3ff353590ba3a3101cbd8b172a3f15551":{"kind":"REWORK","_number":1,"created":"2022-01-10 07:59:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/823938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/38/823938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/1"}}},"commit":{"parents":[{"commit":"ccf6792238db1869911cb457ce9eebfde52f7b49","subject":"Add support for [haproxy_amphora] build_retry_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ccf6792238db1869911cb457ce9eebfde52f7b49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:58:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:58:11.000000000","tz":540},"subject":"Add support for more connection tunables in [haprpoxy_amphora]","message":"Add support for more connection tunables in [haprpoxy_amphora]\n\nThis change adds support for the following tunable parameters in\nthe haproxy_amphora section.\n\n - active_connection_max_retries\n - active_connection_interval\n - faiolver_connection_max_retries\n - faiolver_connection_interval\n\nChange-Id: I6a2686a4e33c75d863ae146a3af37d416e01a655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db2102f3ff353590ba3a3101cbd8b172a3f15551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/db2102f3ff353590ba3a3101cbd8b172a3f15551"}]},"branch":"refs/heads/master"},"a8391b4eaf99274c383b7fdddc8106c430f61da2":{"kind":"REWORK","_number":2,"created":"2022-01-10 08:00:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/823938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/38/823938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/2"}}},"commit":{"parents":[{"commit":"ccf6792238db1869911cb457ce9eebfde52f7b49","subject":"Add support for [haproxy_amphora] build_retry_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ccf6792238db1869911cb457ce9eebfde52f7b49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:58:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 08:00:33.000000000","tz":540},"subject":"Add support for more connection tunables in [haprpoxy_amphora]","message":"Add support for more connection tunables in [haprpoxy_amphora]\n\nThis change adds support for the following tunable parameters in\nthe haproxy_amphora section.\n\n - active_connection_max_retries\n - active_connection_interval\n - faiolver_connection_max_retries\n - faiolver_connection_interval\n\nChange-Id: I6a2686a4e33c75d863ae146a3af37d416e01a655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a8391b4eaf99274c383b7fdddc8106c430f61da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a8391b4eaf99274c383b7fdddc8106c430f61da2"}]},"branch":"refs/heads/master"},"ca3c4cc1c166c473ecf3935936c3ef4946d4c43f":{"kind":"REWORK","_number":3,"created":"2022-01-16 02:05:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/823938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/38/823938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/38/823938/3"}}},"commit":{"parents":[{"commit":"ccf6792238db1869911cb457ce9eebfde52f7b49","subject":"Add support for [haproxy_amphora] build_retry_interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ccf6792238db1869911cb457ce9eebfde52f7b49"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-10 07:58:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-16 02:04:48.000000000","tz":540},"subject":"Add support for more connection tunables in [haprpoxy_amphora]","message":"Add support for more connection tunables in [haprpoxy_amphora]\n\nThis change adds support for the following tunable parameters in\nthe haproxy_amphora section.\n\n - active_connection_max_retries\n - active_connection_retry_interval\n - faiolver_connection_max_retries\n - faiolver_connection_retry_interval\n\nChange-Id: I6a2686a4e33c75d863ae146a3af37d416e01a655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ca3c4cc1c166c473ecf3935936c3ef4946d4c43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ca3c4cc1c166c473ecf3935936c3ef4946d4c43f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
