)]}'
{"id":"openstack%2Fpuppet-octavia~826597","triplet_id":"openstack%2Fpuppet-octavia~master~I4fda43e1742795b98be4f15f960816644900217c","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-11 11:37:58.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-03-07 23:40:15.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4fda43e1742795b98be4f15f960816644900217c","subject":"[health_manager] Allow tuning individual threads","status":"MERGED","created":"2022-01-27 08:21:25.000000000","updated":"2022-03-11 12:02:14.000000000","submitted":"2022-03-11 11:37:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"826597","meta_rev_id":"f7ea97c054b65d5070dbc9b10c484279a7fbea15","_number":826597,"virtual_id_number":826597,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 11:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-03-07 23:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-03-11 10:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-03-11 10:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 13:31:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 16:14:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-03-11 10:02:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"33482dba9e8736a3c808b9defe09d7ef6fa0d459","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 08:21:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a2e2ac54e98409ea894f5461182fc73bc398994","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 10:57:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb604cdda0d1a6e90776f5d663d2f89e74f13d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:31:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eff1cbee97c0432992b290b2df40ccc1 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/690cbd362b5d4558a72b8712df464ef9 : SUCCESS in 3m 38s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f040ba202541d491f7c287e0939fa2 : SUCCESS in 3m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d6f3e536e5d46178860186b5f57eb88 : SUCCESS in 12m 20s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/077561869d0d40889d77cdafd991de82 : SUCCESS in 12m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e39b3b3cd21a411d9e4c79e1ee38bfc8 : SUCCESS in 10m 28s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13071444d55648bb9d253638e70a9df3 : SUCCESS in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aabb9ffb13064fe8af64058537ae562a : SUCCESS in 8m 19s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e96e2b5ca25044f88c6c487ded04e500 : FAILURE in 13m 16s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b84a6495c11d4a8c913a320511ed04e8 : SUCCESS in 27m 46s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/435e0d1fdb2240208d13690f0867ffb6 : FAILURE in 13m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9572e838298c401a9fec505f9467dce7 : FAILURE in 12m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06c9933f96134619a91580ef17b47cdd : SUCCESS in 4m 22s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fca3981cd04b47ea8383ba2ada3cc364 : SUCCESS in 2h 30m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b68828d47b6aaa44cef6ced7a625e1c450795ac2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-27 16:14:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b0b085cf1c5a9bacbaef6998dc4f1e07de5a42e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 17:23:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bfe1d338cefe046ebc90fb076ac256858da248d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 19:45:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/260ebe954ad54b6cbe2efcc35233cdeb : SUCCESS in 4m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e87aa71c60324586bfde1de07b235a65 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24a940b47fcf497b98cc8c4b49e0cc56 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/196fd7e42e9d41f6b43fc1410455970e : SUCCESS in 14m 11s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/288c48704758447cab3b8dccd01c3e4b : SUCCESS in 14m 20s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8560a1f0e02f48d6928ab27fe4bb4fa7 : SUCCESS in 13m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2dc8f93e43f4610af5d2836b8e8285f : SUCCESS in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc90df74c934f358c18851e90aa3ae7 : SUCCESS in 12m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef235059589c48aaae32ae02b6b8a3b9 : FAILURE in 29m 48s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d1fa535e40894b729d3ca469ec2a4dd9 : SUCCESS in 58m 15s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc65fcc53a124b25b30cae21a455ddbb : SUCCESS in 42m 29s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f03e43aa7af0462e8c55b23d8d5d1d15 : SUCCESS in 18m 55s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/c196a886784949999d2a0b298bb5a572 : SUCCESS in 31m 11s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1367ebac1cce4d55bad54e1a7b641853 : FAILURE in 12m 54s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24ac34cf13c44b0a7e7dee5297bb431 : SUCCESS in 21m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51faac61996b4be291b16191176151f2 : SUCCESS in 21m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e11c9de61ab44f73b14af5aa49ca8020 : SUCCESS in 4m 37s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7348d7672b9147fc9f6b0a27d2964f84 : SUCCESS in 1h 56m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bdb00786f89f15accdf102c5491519ea54f56745","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-22 20:46:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"770481a31772497c7bf27919ca500c9a0a9aed3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 03:48:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4601f7618a46d4156d253d130431a3e3e4ae0e18","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 03:49:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a32f6879dcbf7888713346a283ad05a9d1f10989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 05:48:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dee1be2783c4e8c873485ca15d981c3 : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13d9524dad144bd8ac4cbf1278e67899 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c08906eb12154eba87a2dda347954684 : SUCCESS in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/26232d01d0e3426aa214792811056f88 : SUCCESS in 12m 24s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f474c8e7f5d24bc29c33f103d8dc4170 : SUCCESS in 14m 05s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98631c7526c7413488f2233f3dc910f7 : SUCCESS in 8m 45s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ffe8ac7abc94a6d8e4b9d6f862c54f4 : SUCCESS in 8m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b256789d18aa464bb431c67e869d488c : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9d821aaaa264abeb2db5b91303d4aa7 : FAILURE in 24m 08s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/761d83652a564ae7a89ce6d7751de1b7 : SUCCESS in 49m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4d0d48286b64569b4f7f66c034bc7dc : SUCCESS in 46m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87c0555ed13940f08e228055511cca4a : SUCCESS in 17m 08s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/93095fe0419242da8c21ae0de941d67d : SUCCESS in 23m 28s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/16a78f1fbe5c428aa56717a1705bd021 : FAILURE in 9m 47s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce8d45db1a414bd08a969d3b7a44cb82 : SUCCESS in 17m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ebfac289b954a54bbcb17fb6fd045f3 : SUCCESS in 19m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e2fe1609e414116ba72506f3eeef59d : SUCCESS in 3m 17s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6c7da592f06c4a91a86ffc9ce36a8db7 : SUCCESS in 1h 56m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4ba50f3a97a6a4c0ea6f303fe84dd53af2595f3b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:40:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"37bb8e2d0804a90bc36b0d3278a84b93ba250287","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-11 10:02:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e82d7f59f7cca05b1ed790ad29f1cf9b39a89af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 10:10:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d02acb8b5e7b9dbe0a48a1990d7cde7d288024c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:37:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74a17353482144cface12cd9a7fb4fe6 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb9e613bf47b4d5088c2cd28db8ca14c : SUCCESS in 5m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ab5ede52a32460bacf46849cf97d2c7 : SUCCESS in 5m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/266906b86f1f410fac3efe6852246118 : SUCCESS in 14m 47s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/841f9b74f3f7497dba762836e9997e98 : SUCCESS in 14m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49630673a2754da6978ca8e3ea38a5c8 : SUCCESS in 12m 30s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbdd22c971bc4600b62e2844bd90cb80 : SUCCESS in 12m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e042e804848429d9bb3552f16d7ae40 : SUCCESS in 1h 01m 40s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4faceb25ca249dda7d6bb015b43ad55 : SUCCESS in 30m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d16764b78284637987996309e51e0b9 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":4},{"id":"a4a95f79242894d01e6eb71d200b8871fd2d8ff7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 11:37:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f7ea97c054b65d5070dbc9b10c484279a7fbea15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:02:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f5c3f1362d24e6cb0698b8c1b5d5103 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cffdaec79590f5bcac54811a0a06261392e854ee","revisions":{"f1f74762abc32df9a37d0794ff284289f3e1e718":{"kind":"REWORK","_number":1,"created":"2022-01-27 08:21:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/826597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/97/826597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/1"}}},"commit":{"parents":[{"commit":"8f09984e5c3db0f6b302c682021a9f5c0b79d39e","subject":"Merge \"Add support for more connection tunables in [haprpoxy_amphora]\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8f09984e5c3db0f6b302c682021a9f5c0b79d39e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 08:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 08:19:34.000000000","tz":540},"subject":"[health_manager] Allow tuning individual threads","message":"[health_manager] Allow tuning individual threads\n\nThis change allows users to tune health_update_threads and\nstats_update_threads separately, instead of enforcing the same value\nfor these two parameters.\n\nThe workers parameter, to determine the common value for these two\nparameters, has been deprecated in favor of the new separate\nparameters.\n\nChange-Id: I4fda43e1742795b98be4f15f960816644900217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f1f74762abc32df9a37d0794ff284289f3e1e718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f1f74762abc32df9a37d0794ff284289f3e1e718"}]},"branch":"refs/heads/master"},"5c8573b1a7c837907648941f4c1a2551705d86e8":{"kind":"REWORK","_number":2,"created":"2022-01-27 10:57:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/826597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/97/826597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/2"}}},"commit":{"parents":[{"commit":"8f09984e5c3db0f6b302c682021a9f5c0b79d39e","subject":"Merge \"Add support for more connection tunables in [haprpoxy_amphora]\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8f09984e5c3db0f6b302c682021a9f5c0b79d39e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 08:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 10:57:34.000000000","tz":540},"subject":"[health_manager] Allow tuning individual threads","message":"[health_manager] Allow tuning individual threads\n\nThis change allows users to tune health_update_threads and\nstats_update_threads separately, instead of enforcing the same value\nfor these two parameters.\n\nThe workers parameter, to determine the common value for these two\nparameters, has been deprecated in favor of the new separate\nparameters.\n\nChange-Id: I4fda43e1742795b98be4f15f960816644900217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5c8573b1a7c837907648941f4c1a2551705d86e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5c8573b1a7c837907648941f4c1a2551705d86e8"}]},"branch":"refs/heads/master"},"3768b6e6bc6e38edcb095b0966699345ecd69f01":{"kind":"REWORK","_number":3,"created":"2022-02-22 17:23:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/826597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/97/826597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/3"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 08:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 17:23:13.000000000","tz":540},"subject":"[health_manager] Allow tuning individual threads","message":"[health_manager] Allow tuning individual threads\n\nThis change allows users to tune health_update_threads and\nstats_update_threads separately, instead of enforcing the same value\nfor these two parameters.\n\nThe workers parameter, to determine the common value for these two\nparameters, has been deprecated in favor of the new separate\nparameters.\n\nChange-Id: I4fda43e1742795b98be4f15f960816644900217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3768b6e6bc6e38edcb095b0966699345ecd69f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3768b6e6bc6e38edcb095b0966699345ecd69f01"}]},"branch":"refs/heads/master"},"cffdaec79590f5bcac54811a0a06261392e854ee":{"kind":"REWORK","_number":4,"created":"2022-02-23 03:48:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/826597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/97/826597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/97/826597/4"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-27 08:19:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 03:48:24.000000000","tz":0},"subject":"[health_manager] Allow tuning individual threads","message":"[health_manager] Allow tuning individual threads\n\nThis change allows users to tune health_update_threads and\nstats_update_threads separately, instead of enforcing the same value\nfor these two parameters.\n\nThe workers parameter, to determine the common value for these two\nparameters, has been deprecated in favor of the new separate\nparameters.\n\nChange-Id: I4fda43e1742795b98be4f15f960816644900217c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cffdaec79590f5bcac54811a0a06261392e854ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cffdaec79590f5bcac54811a0a06261392e854ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
