)]}'
{"id":"openstack%2Fpuppet-octavia~828311","triplet_id":"openstack%2Fpuppet-octavia~master~I6cdae966ac7bcddbf3510af77e462dec2944ad9c","project":"openstack/puppet-octavia","branch":"master","topic":"amqp_allow_insecure_clients","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-18 09:20:20.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I6cdae966ac7bcddbf3510af77e462dec2944ad9c","subject":"Remove deprecated amqp_allow_insecure_clients","status":"MERGED","created":"2022-02-08 13:17:46.000000000","updated":"2022-02-18 12:33:01.000000000","submitted":"2022-02-18 12:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"828311-amqp_allow_insecure_clients","meta_rev_id":"c88e4223175141cbb9e04a530fb4e8c314b365cd","_number":828311,"virtual_id_number":828311,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-18 12:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-02-09 20:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-02-18 08:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-02-18 08:13:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-08 16:11:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-09 20:13:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-02-18 08:13:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"db4def4752648b11ca837e42d44b9900ff0a3630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:17:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"937c73d21e817aab843ecfc2303b80b190be91f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:18:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"84d816e962888e00c475479a17cdbb3350501c3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-08 13:21:01.000000000","message":"Topic set to amqp_allow_insecure_clients","accounts_in_message":[],"_revision_number":2},{"id":"49654119a45bf3586ac77f8e688c1906752d6da9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 16:11:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5727fc9803b24c2bbca5101cb01d5cff : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cf1848c0c034745a129b8155d0c3db6 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1263bdd9c38142ca8718ccb8eef560cd : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f475b7ab06c40ddb63e6e35cb4de8b3 : SUCCESS in 17m 17s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8412f7e10d9417886aaa299e685d5ab : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aad2a0a809341279d0a79a4ce2721d4 : SUCCESS in 9m 50s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a97428502877400b9dd8afbd53a78a0d : SUCCESS in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ea6254964d48f1899e2e78d7a7979d : SUCCESS in 9m 59s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2e44ed898aa421b9b0e178125702fcb : FAILURE in 14m 01s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffad3038f337484cb62b7313c04a96b0 : SUCCESS in 32m 53s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcfeeb8810634c939521f77af2182801 : FAILURE in 12m 24s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e7b09f5e2cc4fd1a2226d804ab690ca : FAILURE in 18m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4328c694dcd4c4fb2cfcc0ce9598498 : FAILURE in 16m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73f11cb369204ff89f1f0f59eaf7e747 : SUCCESS in 5m 05s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7408143fc6f74e3b9a0d657eed344fc0 : SUCCESS in 2h 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34aeceed3ee802d13601760371a055a93f588337","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-09 20:13:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"83fa8d668bfb5f912f0d64b5f96c566e3fef6d46","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-18 08:13:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"04e301c578c4a55ba4105f3b50763fc2bea248aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 08:15:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4990db5c14e9321410fb7faa24a9408f6919270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 08:56:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a0d23881c6949f9944ab91d65ea8068 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed9f619d11b24654b011c1ae78c2abd2 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ba4ef347294464c9793bbb182cd4f13 : SUCCESS in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4ee5ce665934eb18a07a9b7a86a5ebe : SUCCESS in 11m 23s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f556aa218900404185e1d7bfa4d08146 : SUCCESS in 14m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d5c83c0dae4f5099203530e7ab43d3 : SUCCESS in 12m 47s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/439a94b9972e4870a6f0f28deb5caead : SUCCESS in 12m 06s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d544450dbf344f7b8c9367508cc5f016 : FAILURE in 38m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e2f89339084402cb743707cda780a0e : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"11df915c11a5a4bee20d213042f9265987b7a10a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 09:20:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e70d54443a45fa3a5bce1c41d7e9b7ebe0f622bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-18 09:26:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bfa8e6063173d8c893493664bf98ccd9d738540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:02:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a5d4d7303194fff89341d8e6dee9211 : SUCCESS in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07412da3c2f94f9884ac1015fac45fb9 : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11e368b408b24fa89e89a03fa940a19d : SUCCESS in 2m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/01c612b567f847b98b6f588c2df032b0 : SUCCESS in 12m 06s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/8100da1b3e5e4b6882a05c8c84a3089f : SUCCESS in 11m 50s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d35c7a62b94d59b2a34de3041107b6 : SUCCESS in 7m 56s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37c8fa12766542feb7694db00f44bc06 : SUCCESS in 7m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0cf9d1a54ee458c88c644486530d5a0 : SUCCESS in 8m 07s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02c63b45c2d549ddb0f077669e652b76 : FAILURE in 14m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d31b9ecc0f7f48aa94f08101f71115c6 : SUCCESS in 29m 14s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c4254d580164acdb6d81da667edc016 : FAILURE in 8m 44s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa50c349dc94d67b91ecf3746cd16b8 : FAILURE in 17m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9d88118a7d14c51b0ad8095c940e30f : FAILURE in 29m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fc1c5d947e7447a9e6db9e71a02532c : SUCCESS in 4m 12s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/780bb70fe00b4982a9a5ca4df90434b2 : SUCCESS in 2h 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d97109848f4473e1df08dfe07d4a716550ddc682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:03:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4b31756264c145609acad73574cd6db021995731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:31:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23a10cbc02fb433ea58a8489b5bbb276 : SUCCESS in 3m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b7517383baa40609a7241b45f87d32a : SUCCESS in 3m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2073d4fdb83f437186ab25cd554bd3cd : SUCCESS in 2m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8c8fd6ae93c4bb994a992fcf38011ed : SUCCESS in 13m 21s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/726009c482e142fbaf63ac0df65e1118 : SUCCESS in 8m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a18ed466d9944f659857904f571cfc17 : SUCCESS in 10m 48s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/492bd4bfa339449ab7df67256c2f865a : SUCCESS in 8m 09s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cd037ae42894ecdb6b774339b337a00 : SUCCESS in 26m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06552903e220402e816429b69f7a4dba : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1709f509ccbde49d781fb7ceeb2d6b3a3492b266","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:31:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c88e4223175141cbb9e04a530fb4e8c314b365cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-18 12:33:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da24c244931c42f285cde0db8a556cec : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"93cdf4a66e0b0197766eb5811612fffafd9ad6e6","revisions":{"3be3ce59c6b8f7f058366b66fb24501eaaec8049":{"kind":"REWORK","_number":1,"created":"2022-02-08 13:17:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/828311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/11/828311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/1"}}},"commit":{"parents":[{"commit":"85e5488df2156f2eabb4e9ababe17a9b845d1cec","subject":"Merge \"Clean up deprecated database parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/85e5488df2156f2eabb4e9ababe17a9b845d1cec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:17:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:17:19.000000000","tz":540},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: I6cdae966ac7bcddbf3510af77e462dec2944ad9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3be3ce59c6b8f7f058366b66fb24501eaaec8049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3be3ce59c6b8f7f058366b66fb24501eaaec8049"}]},"branch":"refs/heads/master"},"93cdf4a66e0b0197766eb5811612fffafd9ad6e6":{"kind":"REWORK","_number":2,"created":"2022-02-08 13:18:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/828311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/11/828311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/11/828311/2"}}},"commit":{"parents":[{"commit":"85e5488df2156f2eabb4e9ababe17a9b845d1cec","subject":"Merge \"Clean up deprecated database parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/85e5488df2156f2eabb4e9ababe17a9b845d1cec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:17:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-08 13:18:49.000000000","tz":540},"subject":"Remove deprecated amqp_allow_insecure_clients","message":"Remove deprecated amqp_allow_insecure_clients\n\n... because it was deprecated during Wallaby cycle.\n\nChange-Id: I6cdae966ac7bcddbf3510af77e462dec2944ad9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/93cdf4a66e0b0197766eb5811612fffafd9ad6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/93cdf4a66e0b0197766eb5811612fffafd9ad6e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
