)]}'
{"id":"openstack%2Fpuppet-octavia~830317","triplet_id":"openstack%2Fpuppet-octavia~master~I20e91a095b0424f21c66785f6d27a0d3ecaea0eb","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-01 08:55:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20e91a095b0424f21c66785f6d27a0d3ecaea0eb","subject":"Make TripleO job voting","status":"MERGED","created":"2022-02-22 07:05:17.000000000","updated":"2022-08-01 08:55:22.000000000","submitted":"2022-08-01 08:55:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830317","meta_rev_id":"7fcc13ace0ac342dcebf91a1f300ee36fa0812f9","_number":830317,"virtual_id_number":830317,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 08:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-01 07:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-07-12 04:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2022-08-01 07:33:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 09:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 14:57:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-01 07:33:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"0d8060a90e0a7981f51290d43ab8c3a37520a4d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 07:05:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30b13fb63c2a15cdbb3df889f4b404fd7c20844d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 09:09:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a924153e074e4cc9b7a6cd60cea6adb4 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15ad57e9cc514c64b84bc5e4ef40ab91 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c969e600368408ab816af10a5fa54e9 : SUCCESS in 3m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c56d338982ae4b70a6f448ad837a8d81 : SUCCESS in 13m 44s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d717631c1f94992ab6df7dd5a3b2703 : SUCCESS in 13m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eace491033b14138ba21fe4a8b2b9ff8 : SUCCESS in 9m 31s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e9489ea92f94e25b1613fe828025d5c : SUCCESS in 9m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaffe591b78e4966ab6129b132cdf784 : SUCCESS in 9m 51s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9152aa76b86840feb42a5869a531b728 : FAILURE in 30m 29s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f5b663008cb45e3ab688438222337a6 : SUCCESS in 1h 19m 18s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1704f638ee549a69c2564ed435ab273 : SUCCESS in 56m 56s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab26e7caf184273ab7c9ec59b91617f : SUCCESS in 19m 47s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/947616fc4fc74e5a83ea27ac1ba5e2ce : SUCCESS in 29m 16s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c6442aac6c14992a7a4474b37f758d0 : FAILURE in 12m 01s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c753c97d9bc24c6f8e5356a5c5983b87 : SUCCESS in 20m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/099b5bced14540459f4b0943de7c2857 : SUCCESS in 20m 09s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4f6dc463b85841dd8275c4977ad9adfb : FAILURE in 2h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"54a7b3d4edd90f88e5491640f41497b0acc5bf22","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-22 10:27:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16bd918177a89abe67a7c607c138b57748970f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 12:28:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0125a76f70c14e549461761776b075c2 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/280e8ea7681349b6a1e19b91818e6010 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a24977983fe4f50b772327fc1707d7e : SUCCESS in 3m 34s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/061c9206eb864a3f81034026325ba640 : SUCCESS in 12m 02s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/43a17d65c53544b2a8399dd4afdf4afa : SUCCESS in 12m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9cb69ba5dff414faa81e13786660bdd : SUCCESS in 10m 11s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcd90764f9284051bc189c05930f29b0 : SUCCESS in 10m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/decbb17c7d624ac58f33d523dd369f80 : SUCCESS in 9m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eaaa42be5744a36bed828bb85f05c77 : FAILURE in 25m 40s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f610471c8fc4fd9b74e9034776358b8 : SUCCESS in 1h 13m 02s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e625d29c25424a548bbc8632be9afd3d : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e8ed092fc4452bb70b3251de367ae3 : SUCCESS in 22m 49s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/821a642a106c4627a4128a633ff3e7fb : SUCCESS in 27m 06s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/53bf4c8c74224e149ee9d98e691f01e8 : FAILURE in 8m 41s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/252bd7e6ecc64c50ba84e7c9ab7717eb : SUCCESS in 21m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/178524f5b74545b2b4323d5cd1e80576 : SUCCESS in 22m 11s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/256d98e149bc4ed28ec223b6eb9aad6b : SUCCESS in 1h 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4de02446c5585a68bdac8fc03ac031693878baac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-02-22 14:57:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae4ec414cc333bdbaee877e4df49991cc80b08ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-12 04:43:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"71ba7e871575e3183bc035c0b3e0ec43d076ea3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 06:40:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/940ae660bed24394892f411d8d3eefaa\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/512d725d6b91435aa946c6c13e89c465 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c4c91d522e342f3b5557ea4903fcc1b : SUCCESS in 3m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcfb4cae13454623976df625126ae634 : SUCCESS in 3m 18s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/be3cd3250cc04707a9921b01c9e3116f : SUCCESS in 15m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a5e14fdd7a1466081920bc8403f03cd : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fab8a1bc8f644d31b62cc78337f5bb97 : SUCCESS in 9m 25s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3da3bc02884dc2920d5ba9229d4dca : SUCCESS in 10m 07s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20388ea842ed486f9c498effceced2f2 : FAILURE in 23m 52s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/44d890e751884582b0092c5021f56f47 : SUCCESS in 54m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/284b68f62fa34c11a14ab55f74fcb6c7 : SUCCESS in 18m 59s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f55160b3c73d4102bc15a3c6bbd9df69 : SUCCESS in 28m 18s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f0e3aa5deed4bfda4b54714f45f7ba3 : SUCCESS in 18m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52e2e14c7c9e4ca6b0db02d1f05ee77f : SUCCESS in 21m 29s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dc7144963d9a4a1883287309c44c1dfc : SUCCESS in 1h 52m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0f3c4dc31b670b9d041a2a6cfba76f7e3c39ed2d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-01 07:33:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e7ab97264fb5401d25c4ca3de03bc2aad467c44f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:35:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11861b37a791aa6be7f8cc99f1bb22518c7659cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 08:55:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddf144865e49452495e74caa90d2206d\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8832f7f5afae45afaf019483e5340019 : SUCCESS in 4m 21s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/988d7811c1da4f1eb04805dcc7b39c8c : SUCCESS in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50698f20014841f1a62fd49b2deebb84 : SUCCESS in 3m 35s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee4369d610cd4e509450f22a028d4ad8 : SUCCESS in 16m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fddc0b10130f4e0fa93841917f4b9e64 : SUCCESS in 11m 22s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ea01c7f36544adeaf68a42fc9132b16 : SUCCESS in 11m 46s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c010e084c7ae454f826fa7240f5332db : SUCCESS in 54m 56s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d82ac471fa7455dbff3785042de2a90 : SUCCESS in 33m 07s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e1ba684e6adc492daae0fd27adbec439 : SUCCESS in 1h 13m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7fcc13ace0ac342dcebf91a1f300ee36fa0812f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 08:55:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a91f1e7a69d98a072fdcc9eb7af9c018bbbfc08","revisions":{"2220b772dceec7e2bd082a393194176ab192742b":{"kind":"REWORK","_number":1,"created":"2022-02-22 07:05:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/830317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/17/830317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/1"}}},"commit":{"parents":[{"commit":"7edbd6dcae7bca21e3e623828803c74d9bd36c07","subject":"Merge \"api: Add support for TLS cipher/version parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/7edbd6dcae7bca21e3e623828803c74d9bd36c07"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 07:03:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 07:05:06.000000000","tz":540},"subject":"Make TripleO job voting","message":"Make TripleO job voting\n\nThe puppet-octavia-tripleo-standalone job has been stable since we\nswitched to nested virt. Because this is the only test case which\nvalidates actual resource operations (e.g. load balancer management,\npool management, ...), let\u0027s enable the job to ensure any change does\nnot affect functionality.\n\nChange-Id: I20e91a095b0424f21c66785f6d27a0d3ecaea0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2220b772dceec7e2bd082a393194176ab192742b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2220b772dceec7e2bd082a393194176ab192742b"}]},"branch":"refs/heads/master"},"1a91f1e7a69d98a072fdcc9eb7af9c018bbbfc08":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-12 04:43:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/830317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/17/830317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/17/830317/2"}}},"commit":{"parents":[{"commit":"aa83e0b86fccb0572f8ca17785cc2e8542ff5d1c","subject":"Merge \"Fix typos in parameter descriptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/aa83e0b86fccb0572f8ca17785cc2e8542ff5d1c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-22 07:03:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-12 04:43:04.000000000","tz":0},"subject":"Make TripleO job voting","message":"Make TripleO job voting\n\nThe puppet-octavia-tripleo-standalone job has been stable since we\nswitched to nested virt. Because this is the only test case which\nvalidates actual resource operations (e.g. load balancer management,\npool management, ...), let\u0027s enable the job to ensure any change does\nnot affect functionality.\n\nChange-Id: I20e91a095b0424f21c66785f6d27a0d3ecaea0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1a91f1e7a69d98a072fdcc9eb7af9c018bbbfc08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1a91f1e7a69d98a072fdcc9eb7af9c018bbbfc08"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
