)]}'
{"id":"openstack%2Fpuppet-octavia~830567","triplet_id":"openstack%2Fpuppet-octavia~master~Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2","project":"openstack/puppet-octavia","branch":"master","topic":"provider_drivers","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-08 01:07:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2","subject":"Accept array or hash for enable_provider_drivers","status":"MERGED","created":"2022-02-23 09:53:10.000000000","updated":"2022-03-08 01:10:44.000000000","submitted":"2022-03-08 01:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830567-provider_drivers","meta_rev_id":"4847be681e8ad8c87aa87481d106a6bdd974d65c","_number":830567,"virtual_id_number":830567,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 01:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-24 09:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-07 23:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-07 23:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 15:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 09:14:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-07 23:40:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"fd30a28ed5557b55ac9170eb28a08a6fbae2c64f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 09:53:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49b54514f72edaaa40bf6202eb5acac85feffc8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 09:59:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"38b05bfc67932bdc9a2610aaa82479779367d645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 10:01:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14503f17f6eb185b0dad26887966b91fd7bf76de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 10:47:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"66bd60596462c1542b4b7a1dd39fb7a15897d361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 11:07:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7095b3d717df06c3e86a4f4f12843250c1c349aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 11:26:48.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d1cacf7d2f572dd512845cdd1f6ffe2b2c076db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 13:30:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"54bbb633c2db74451a0227706e94b18920306b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 15:35:42.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92479713e37b4cfa9770747680ca1e9a : SUCCESS in 3m 25s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1e0501830af4e6f9682c853e3be7dc2 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/131d0061374d46ff9bd40abfa81150f6 : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fdd69fdd8644d2299175d3d57e4f5c7 : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd4a9efe98dd4fefba75b95c53245296 : SUCCESS in 13m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbf24fa0e2534c53a380ff1a575ad559 : SUCCESS in 11m 54s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a134f64c6a4a63980d97f54429d914 : SUCCESS in 9m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d183d1ef18c04c49ae3b733ffa9623e2 : SUCCESS in 10m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40fec3d605ae440083dee28465e61aa1 : FAILURE in 27m 19s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc1352b838c94398a4c16b146571bb66 : SUCCESS in 1h 01m 14s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3da8b94fcf074a1f8394d487b18dee77 : SUCCESS in 46m 07s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f0ac9c61424566b61f8117b753203a : SUCCESS in 19m 42s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/795ef1a3dc79463ca8c20767e948f4c1 : SUCCESS in 27m 59s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/31197c8218bb4a268107e4793bb8a215 : FAILURE in 8m 34s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36aa855b864745e895cabd3f06d98824 : SUCCESS in 19m 36s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff744958e7b74d408f65b8a595aa8e3c : SUCCESS in 19m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f401205b386348669a8e1965a64477ff : SUCCESS in 3m 45s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f98a4325e57c4e96bc9c12280480a980 : SUCCESS in 2h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"180d10d238741782f7d43a3a48da8e65e4f514ee","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-24 09:14:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"f1233ef515ddc178510de76c31b8b403f85e4ae2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:40:26.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"426993d595921d43b555efdd72b8012a601214b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:42:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f92fc1c21342f1a598d4f7b7ec5bf73e352f16dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:01.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5778a58835e54b41bbd59bbac1425b3c : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a96d2987d5294fad8b69c96665d393bc : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/178dab23c57b47a88b6f02681b4960fa : SUCCESS in 4m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/32705df341104cb099675ab053cf65ee : SUCCESS in 14m 35s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b63ab99ca0be4193a8aa83c66eed5a3a : SUCCESS in 15m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c283d98488d544838e5550e9ef39566c : SUCCESS in 10m 44s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e11b87032e1453384315f4ed813c757 : SUCCESS in 11m 36s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/48b686a737284b4f97cb49785197dd5b : SUCCESS in 1h 10m 43s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3f72b7b4a0a4226ad80449ddd0e8fce : SUCCESS in 28m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b282f022e0948f7b0eecd66247cf803 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":7},{"id":"70e4f4ec437ed4bf8f31cc54f628d5d0e21a88b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4847be681e8ad8c87aa87481d106a6bdd974d65c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:10:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96c59951acbf4dad97857ac1f794d328 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"270d33d5436b0e9f13f88b308a17e299e00f21bc","revisions":{"8d4765d44f32704904637afe5fc61701b94368f1":{"kind":"REWORK","_number":1,"created":"2022-02-23 09:53:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/1"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:52:45.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [DEFAULT] enable_provider_drivers parameter is DictOpt and accepts\na string value which represents a dictionary. This change allows using\narray or has for this parameter so that users can use more \"native\"\ntype to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8d4765d44f32704904637afe5fc61701b94368f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/8d4765d44f32704904637afe5fc61701b94368f1"}]},"branch":"refs/heads/master"},"5f0719027fe3ea65f63b27f1358dd421abf8b815":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-02-23 09:59:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/2"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:58:13.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f0719027fe3ea65f63b27f1358dd421abf8b815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f0719027fe3ea65f63b27f1358dd421abf8b815"}]},"branch":"refs/heads/master"},"24d71e176fa45c58922d1647eb2b16a19d9891ec":{"kind":"REWORK","_number":3,"created":"2022-02-23 10:01:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/3"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 10:01:08.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/24d71e176fa45c58922d1647eb2b16a19d9891ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/24d71e176fa45c58922d1647eb2b16a19d9891ec"}]},"branch":"refs/heads/master"},"09b3a3c3d7106d1d3394be084998ef6ca13c1f33":{"kind":"REWORK","_number":4,"created":"2022-02-23 10:47:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/4"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 10:46:27.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/09b3a3c3d7106d1d3394be084998ef6ca13c1f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/09b3a3c3d7106d1d3394be084998ef6ca13c1f33"}]},"branch":"refs/heads/master"},"99d567d8219084567b08876ff7671f713d96fb40":{"kind":"REWORK","_number":5,"created":"2022-02-23 11:07:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/5"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:07:02.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/99d567d8219084567b08876ff7671f713d96fb40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/99d567d8219084567b08876ff7671f713d96fb40"}]},"branch":"refs/heads/master"},"17193782120df6d17099db5ee9dadb4c428d6b00":{"kind":"REWORK","_number":6,"created":"2022-02-23 11:26:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/6"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:26:01.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17193782120df6d17099db5ee9dadb4c428d6b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17193782120df6d17099db5ee9dadb4c428d6b00"}]},"branch":"refs/heads/master"},"270d33d5436b0e9f13f88b308a17e299e00f21bc":{"kind":"REWORK","_number":7,"created":"2022-02-23 13:30:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/830567/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/67/830567/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/67/830567/7"}}},"commit":{"parents":[{"commit":"39e9c32801a1bc1caadb0904cd94145a6b1cb9d8","subject":"Merge \"Enable scenario test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/39e9c32801a1bc1caadb0904cd94145a6b1cb9d8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 13:30:44.000000000","tz":540},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/270d33d5436b0e9f13f88b308a17e299e00f21bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/270d33d5436b0e9f13f88b308a17e299e00f21bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
