)]}'
{"id":"openstack%2Fpuppet-octavia~830590","triplet_id":"openstack%2Fpuppet-octavia~master~Id0371d4f9dd4a4922deabae512153d9f9767bf55","project":"openstack/puppet-octavia","branch":"master","topic":"provider_drivers","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-08 01:07:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0371d4f9dd4a4922deabae512153d9f9767bf55","subject":"Use consistent name for enabled_provider_drivers","status":"MERGED","created":"2022-02-23 09:59:49.000000000","updated":"2022-03-08 01:07:04.000000000","submitted":"2022-03-08 01:07:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830590-provider_drivers","meta_rev_id":"215ab74be33dba775551921ad416227b0e297f23","_number":830590,"virtual_id_number":830590,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 01:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-02-24 09:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-03-07 23:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-03-07 23:40:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-23 16:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-24 09:14:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-03-07 23:40:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c55bd91006cc6a6c0ba4166b8e5bfa5fa99c8445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 09:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"795d3897ac716b596de592ef745e9ada861b3f21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 10:01:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b32c1dd2ac22b09a99bb37ac263db05b862a2d72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 10:47:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0f2f684a18275d3b2e3853024721bf2ffcd4736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 11:07:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"33082cccb631bcabed63b3c4a74a218a6d3f08a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 11:26:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"12a3ed7c96e632515aef1335d219c951c9e1d0e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 13:30:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d65f6f7c94c6ff8cdc6fec92c23cae655f13df0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-23 14:05:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"aa7dcc6fafe3d0b918c458addb9f92db6f61c839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 16:16:29.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8c9a254f2ad415ea8e823cc0c8a1d09 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a1e491d3644fb5813936c1085682e6 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813f02c682674d23927e8ffe1b633bd4 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f185af24c0d94ced980c1b44ab39fa45 : SUCCESS in 13m 59s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/938b2656ef9940a2b2c11f6f027dd700 : SUCCESS in 12m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/354981e153c7499b91bc13281e959d0e : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f4ce3b9a07d4a0fa4484fff0968e874 : SUCCESS in 10m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1610056edd814237ad646ddc2a378b7d : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cfb796b8eef415aa60f570330c21655 : FAILURE in 31m 22s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ebab51bb0a654a3a89da09b6a737b76b : SUCCESS in 55m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b583396e635d4a7fa014888d95044679 : SUCCESS in 1h 03m 37s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fbd0f3409144ce4bfa9fc5a95f308b1 : SUCCESS in 19m 11s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d9ccf3681cc40cc99d59b2da9d5c521 : SUCCESS in 31m 38s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec41118470b84fc496c8823fbb7bea9d : FAILURE in 9m 15s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9eadf754d60e4c31b3d2090c8c836722 : SUCCESS in 21m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48ba7569451d4141850e69e9295b6c03 : SUCCESS in 22m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db545e91186b4e039d69fee0698f4257 : SUCCESS in 3m 53s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2fe7c342421545d68874ef59ddd1658d : SUCCESS in 2h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"099a4c2c0fb1f012cc5cc3bbaff13d701e7e1345","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-24 09:14:55.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0b2dca92cae5b742a0648c77dc25f24b3e9fae92","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:40:32.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a71d6766d20f152f5e7ffb4582fcc2c23103fbae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:42:10.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c3adb3ed6d3a3841671e7a9ae3ec3e43201d818e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:04.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52c872a26d814f28a3552d0f199546c3 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5fb36485be7406babdd96d453b7f308 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c09e5f8a862a4f24bf80d0017c229dd0 : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/002bcd4b3d7542538b4c67b83bfa8d08 : SUCCESS in 15m 50s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/07ae33ca72df4485b294a1c5dba0f545 : SUCCESS in 12m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c56ec4744a1441e39717d1c73539ed63 : SUCCESS in 13m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/909306057f9d476fa2e915dec2f41e39 : SUCCESS in 10m 51s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9514ad1b4f84ad98abd70c57169aa88 : SUCCESS in 56m 27s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c117d3e792b48c2abc72a80504c4b93 : SUCCESS in 30m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/032a630c37354bb38c505a2e0e2bdd2c : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":7},{"id":"215ab74be33dba775551921ad416227b0e297f23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2b9282bab65d879b47c4a1065801e0e9d916ecad","revisions":{"c7990c4d911d99010adfdb299f6e6a11910b6793":{"kind":"REWORK","_number":1,"created":"2022-02-23 09:59:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/1"}}},"commit":{"parents":[{"commit":"5f0719027fe3ea65f63b27f1358dd421abf8b815","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/5f0719027fe3ea65f63b27f1358dd421abf8b815"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:59:34.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c7990c4d911d99010adfdb299f6e6a11910b6793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c7990c4d911d99010adfdb299f6e6a11910b6793"}]},"branch":"refs/heads/master"},"231a819e4757a7c57b6e004c829453ec2aa966d5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-02-23 10:01:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/2"}}},"commit":{"parents":[{"commit":"24d71e176fa45c58922d1647eb2b16a19d9891ec","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/24d71e176fa45c58922d1647eb2b16a19d9891ec"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 10:01:16.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/231a819e4757a7c57b6e004c829453ec2aa966d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/231a819e4757a7c57b6e004c829453ec2aa966d5"}]},"branch":"refs/heads/master"},"e96466eed8d2187a2801c75d8de7d845e4d03861":{"kind":"REWORK","_number":3,"created":"2022-02-23 10:47:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/3"}}},"commit":{"parents":[{"commit":"09b3a3c3d7106d1d3394be084998ef6ca13c1f33","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/09b3a3c3d7106d1d3394be084998ef6ca13c1f33"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 10:47:46.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e96466eed8d2187a2801c75d8de7d845e4d03861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/e96466eed8d2187a2801c75d8de7d845e4d03861"}]},"branch":"refs/heads/master"},"b9b9c3872c3e4c77842d09203fa3c16f04f431a8":{"kind":"REWORK","_number":4,"created":"2022-02-23 11:07:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/4"}}},"commit":{"parents":[{"commit":"99d567d8219084567b08876ff7671f713d96fb40","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/99d567d8219084567b08876ff7671f713d96fb40"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:07:33.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b9b9c3872c3e4c77842d09203fa3c16f04f431a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b9b9c3872c3e4c77842d09203fa3c16f04f431a8"}]},"branch":"refs/heads/master"},"d2d2bd8b8d0c6d9838cb9c46cd3a156d38c09910":{"kind":"REWORK","_number":5,"created":"2022-02-23 11:26:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/5"}}},"commit":{"parents":[{"commit":"17193782120df6d17099db5ee9dadb4c428d6b00","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/17193782120df6d17099db5ee9dadb4c428d6b00"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 11:26:36.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d2d2bd8b8d0c6d9838cb9c46cd3a156d38c09910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d2d2bd8b8d0c6d9838cb9c46cd3a156d38c09910"}]},"branch":"refs/heads/master"},"f626c56c9d33a7930e831e8c680437721f2916ae":{"kind":"REWORK","_number":6,"created":"2022-02-23 13:30:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/6"}}},"commit":{"parents":[{"commit":"270d33d5436b0e9f13f88b308a17e299e00f21bc","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/270d33d5436b0e9f13f88b308a17e299e00f21bc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 13:30:47.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f626c56c9d33a7930e831e8c680437721f2916ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f626c56c9d33a7930e831e8c680437721f2916ae"}]},"branch":"refs/heads/master"},"2b9282bab65d879b47c4a1065801e0e9d916ecad":{"kind":"REWORK","_number":7,"created":"2022-02-23 14:05:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/830590/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/90/830590/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/90/830590/7"}}},"commit":{"parents":[{"commit":"270d33d5436b0e9f13f88b308a17e299e00f21bc","subject":"Accept array or hash for enable_provider_drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/270d33d5436b0e9f13f88b308a17e299e00f21bc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:57:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 14:05:23.000000000","tz":540},"subject":"Use consistent name for enabled_provider_drivers","message":"Use consistent name for enabled_provider_drivers\n\nThis change renames the octavia::api::provider_drivers parameter so\nthat the name is consistent with the parameter name in Octavia.\n\nChange-Id: Id0371d4f9dd4a4922deabae512153d9f9767bf55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b9282bab65d879b47c4a1065801e0e9d916ecad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b9282bab65d879b47c4a1065801e0e9d916ecad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
