)]}'
{"id":"openstack%2Fpuppet-octavia~831504","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~I7424536c93e5bfa13d5b4bb5a79ded5b5845d5f6","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"c9_wallaby_gates","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-03-08 01:07:07.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-02 15:06:13.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I7424536c93e5bfa13d5b4bb5a79ded5b5845d5f6","subject":"Add centos-9 jobs for wallaby zuul layout","status":"MERGED","created":"2022-03-02 10:19:48.000000000","updated":"2022-03-08 01:07:07.000000000","submitted":"2022-03-08 01:07:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831504-c9_wallaby_gates","meta_rev_id":"ddfa8af51bb61bc9d99e5a8ef5fd9aca4fda0fba","_number":831504,"virtual_id_number":831504,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-08 01:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-03-04 09:03:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-03-02 15:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-03-07 23:40:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-03-07 23:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 10:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 10:32:36.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-03-07 23:40:36.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3bf36763d06ee59f9d641943c47bf0474e401a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-02 10:19:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7cd7a66eb7a048fd67510a5cb7289addbc3ee6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:20:43.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/puppet-octavia on branch stable/wallaby.  The error was:\n\n  while parsing a block collection\n  in \"openstack/puppet-octavia/.zuul.yaml@stable/wallaby\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/puppet-octavia/.zuul.yaml@stable/wallaby\", line 25, column 2","accounts_in_message":[],"_revision_number":1},{"id":"64bc822d4c160fed4dd16d56a076286b3dbbbb38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-02 10:22:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c1427a85aa7b58271bcb22d480c276a9f5dc173c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-02 10:23:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"15eb3afcfe844a751a2b51e7e3af33af59a154a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:23:32.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/puppet-octavia on branch stable/wallaby.  The error was:\n\n  while parsing a block collection\n  in \"openstack/puppet-octavia/.zuul.yaml@stable/wallaby\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/puppet-octavia/.zuul.yaml@stable/wallaby\", line 25, column 2","accounts_in_message":[],"_revision_number":2},{"id":"d372cf1b11e93378684602576dc54dc853061f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:07:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e6ebed01ddd48e4ab1f5fe1e63c76c0 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/799e742d90414d93ae4164e5d3aa34bc : SUCCESS in 5m 52s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3553d6710d5d4004929d5516a5052000 : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/129d55c6d8284c329cbd66bc77dab310 : SUCCESS in 10m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9a901148556408799cb2b3598f7d99a : SUCCESS in 10m 56s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/29e47a469af744d3a0fb6810255b3f80 : SUCCESS in 30m 44s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/169c94db61f144629a5da2dfc8baa768 : FAILURE in 18m 14s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/49cfcef4144d4c07867ad6d429b4bfcb : SUCCESS in 2h 38m 40s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/257c159fab344515bd6a7acee76bf0ca : SUCCESS in 1h 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16ccdc58bf76cdd708e0c55b8c2139ad85bcecef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-03-02 15:06:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3fc746317dc5b82709b1ccecf63845e354788ca7","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-03-04 09:03:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c04ecd6354589d26ae0a2539ec2e09b622c1536d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-03-07 23:40:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"972d30e66e7e32ecbd9c68b872a77f2ab917dcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 23:42:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70930721b716459902d394e552036defeb07cd26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46abedd2bbf041d19dd5099034463b75 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f09833c3d3f94336b58a1cf20afcfc8a : SUCCESS in 5m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7681802d7e54b1580de145d6720a1fd : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdcb135087a649b3a4d9073150c95062 : SUCCESS in 13m 28s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8556e725ab848fa981b23a327c1db8c : SUCCESS in 32m 48s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8579b68e175c4c2786be0823dcd2a316 : FAILURE in 18m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddfa8af51bb61bc9d99e5a8ef5fd9aca4fda0fba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 01:07:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2255949c3d9a0d50c53e931437a368106f53be1b","revisions":{"bc06050152d9f9099e70c3d13ecacce9f8b3032a":{"kind":"REWORK","_number":1,"created":"2022-03-02 10:19:48.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/04/831504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/04/831504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/1"}}},"commit":{"parents":[{"commit":"da1dcd49927e55c3951b8da7c29d384e1e081363","subject":"api: Add support for TLS cipher/version parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/da1dcd49927e55c3951b8da7c29d384e1e081363"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:19:08.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:19:08.000000000","tz":120},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-octavia-tripleo-centos-9-standalone definition\nand into the check zuul layout. This is added as non voting\nas puppet openstack will not support c9 on wallaby.\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: I7424536c93e5bfa13d5b4bb5a79ded5b5845d5f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bc06050152d9f9099e70c3d13ecacce9f8b3032a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bc06050152d9f9099e70c3d13ecacce9f8b3032a"}]},"branch":"refs/heads/stable/wallaby"},"63ad932ab2d55a627070f82d67d991fd9ba15c70":{"kind":"REWORK","_number":2,"created":"2022-03-02 10:22:43.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/04/831504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/04/831504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/2"}}},"commit":{"parents":[{"commit":"da1dcd49927e55c3951b8da7c29d384e1e081363","subject":"api: Add support for TLS cipher/version parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/da1dcd49927e55c3951b8da7c29d384e1e081363"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:19:08.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:22:43.000000000","tz":0},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-octavia-tripleo-centos-9-standalone definition\nand into the check zuul layout. This is added as non voting\nas puppet openstack will not support c9 on wallaby.\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: I7424536c93e5bfa13d5b4bb5a79ded5b5845d5f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/63ad932ab2d55a627070f82d67d991fd9ba15c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/63ad932ab2d55a627070f82d67d991fd9ba15c70"}]},"branch":"refs/heads/stable/wallaby"},"2255949c3d9a0d50c53e931437a368106f53be1b":{"kind":"REWORK","_number":3,"created":"2022-03-02 10:23:21.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/04/831504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/04/831504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/04/831504/3"}}},"commit":{"parents":[{"commit":"da1dcd49927e55c3951b8da7c29d384e1e081363","subject":"api: Add support for TLS cipher/version parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/da1dcd49927e55c3951b8da7c29d384e1e081363"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:19:08.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-03-02 10:23:20.000000000","tz":0},"subject":"Add centos-9 jobs for wallaby zuul layout","message":"Add centos-9 jobs for wallaby zuul layout\n\nAdds puppet-octavia-tripleo-centos-9-standalone definition\nand into the check zuul layout. This is added as non voting\nas puppet openstack will not support c9 on wallaby.\nSee the topic branch [1] for related reviews and [2] for context.\n\n[1] https://review.opendev.org/q/topic:c9_wallaby_gates\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2022-February/027403.html\n\nDepends-On: https://review.opendev.org/c/openstack/puppet-openstack-integration/+/831206\nChange-Id: I7424536c93e5bfa13d5b4bb5a79ded5b5845d5f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2255949c3d9a0d50c53e931437a368106f53be1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2255949c3d9a0d50c53e931437a368106f53be1b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
