)]}'
{"id":"openstack%2Fpuppet-octavia~836019","triplet_id":"openstack%2Fpuppet-octavia~master~Icdee753af276224c969a3eb0e790d331269893d6","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-04 21:07:19.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-04-04 20:03:28.000000000","reason":"Tobias Urdin replied on the change"}},"hashtags":[],"change_id":"Icdee753af276224c969a3eb0e790d331269893d6","subject":"Add allow_ping_health_monitors parameter to api","status":"MERGED","created":"2022-03-31 13:16:28.000000000","updated":"2022-04-04 21:08:32.000000000","submitted":"2022-04-04 21:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836019","meta_rev_id":"7da86a9da19d184d52c5ddf6546494b58b8af030","_number":836019,"virtual_id_number":836019,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 21:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-04-01 01:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-04-04 20:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-04-04 20:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 14:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 01:02:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-04-01 01:04:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"5fed321d64373372da3dc99f78bbbe05f25c67cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-31 13:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd2742259ba514008452fd10d4a06de6f2c357a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:34:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff3f044325640da860f0cd1f20e1cc0 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41bd85b27319493cb89c16636001ddc3 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b195952a55054d3e9b0b9593f281e4f4 : SUCCESS in 3m 14s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c067651d5e3146d0b2c3595395e9b66c : SUCCESS in 11m 59s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/23917eca928c4e6b93dc1495a8c13868 : SUCCESS in 12m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6522e2709e784a538680e7861bcb01b3 : SUCCESS in 10m 04s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb4f56ea181d45899d980dcc42724392 : SUCCESS in 10m 45s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8708399e732745d8ba0ec1e3c6f9e6bf : SUCCESS in 9m 41s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50a6404e26fa4ea3b47208aee567adf3 : FAILURE in 41m 45s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c186f4c10f94d0ab0833053982bcdbd : SUCCESS in 1h 05m 50s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/97f6b77b08824a8d84afb91da080ca8a : SUCCESS in 49m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39d5e1afa843466d9b3a7e192e0e7ff9 : SUCCESS in 20m 22s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4420cdc18d548ebb56eae13b0b24e60 : SUCCESS in 27m 05s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/09c77edab0b5476b989a0ed3a08d8c65 : NODE_FAILURE in 0s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4fbd683650cb489fb483b901f5b16d1c : SUCCESS in 19m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01789e1e86c449df8e72701bc787e5ab : SUCCESS in 21m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246eae03d68445548eb67bbe7ffc689f : SUCCESS in 3m 49s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/940e2d0499e548dca6c9347c76a21810 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3352cb03293ec52740edb32edf8a883079ec05b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-01 01:02:55.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2b569c122cd83d93f12d056b88d4299d60eab196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-01 01:04:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"92355d93d65e68834570553166018de9d9c9e0ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-04-01 01:04:19.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6a0d8de42a162b65ed9e04774b8b6c0039d8c454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 02:09:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc51a8e0f14f4fbeaf61fc251bd30761 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622a2aea34214b4ea745b4a626b11374 : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc2086e583d4e319c64a1760afac3c3 : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/41868f0cbc864045a24fba28e633d9f2 : SUCCESS in 10m 52s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9bfaf7055574d31ac85264b182cc713 : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ef5658edbd45f0a0cd005218d31aa6 : SUCCESS in 9m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0494424fc537459e9aa4cb80794c2267 : SUCCESS in 8m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a4731c4b7594e5694c02c1ab2d22ff1 : SUCCESS in 8m 35s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7925286a038e4f1ca8d2b685116b0df8 : FAILURE in 33m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b25b195cb6d4419810e87aa8fed53df : SUCCESS in 1h 03m 03s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/548d620ca3e345eda521daa70cd4e07a : SUCCESS in 40m 14s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38ade268d8aa46b6b560fe5053716469 : SUCCESS in 19m 50s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/0416897f5a594897a3d9fb6c34f84262 : SUCCESS in 24m 25s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd69c3f0e8744046b3c17ce30bf6a030 : FAILURE in 7m 12s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac7631653074495a06ba5865cef102c : SUCCESS in 18m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c63269a6fd648ff93dfbc9ed076c31e : SUCCESS in 18m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5c4e1fa2890402683c392d4d8236728 : SUCCESS in 3m 31s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9746acb1fef944239f111f452b04e808 : FAILURE in 7m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b83d68f16cbf005b6c93c3ebb53ab84188658739","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-04-04 20:03:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed0e7a030b6935c56ec1a4bfc149737a83c94165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 20:04:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ccf16d4749b366fcfcd351021e93ab066df4bdda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 21:07:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaf2f9a5a8424fa4975aedae2e856e6a : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c56aef1ee9462eba333dfbe8948b3e : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3c6aa32ff7b4ba48b59e6cd686080c8 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/56860103540e42eab73800239233e25b : SUCCESS in 11m 54s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/11226598ac37490189ccabc881eccc9c : SUCCESS in 12m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a5c890f47b24d28b852b4b8d58008e4 : SUCCESS in 9m 07s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24dba396f7b14810bb4343aa40481698 : SUCCESS in 10m 12s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/085d95e3199c4280a0f364d607132939 : SUCCESS in 58m 16s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5392a6f4fd124132b533102dc0c994dd : SUCCESS in 34m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bea822e0f09b4df29ee7b1efbacd8a36 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"af9178c033033279bf0c7a5658013735d60dcb83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 21:07:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7da86a9da19d184d52c5ddf6546494b58b8af030","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 21:08:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e53e370fdeb44df3a17ded8090cd1e93 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c3d78e5cb45d0a5d5c360746c2c06298b638ec57","revisions":{"ae2f83a3a54dd17bd45831ad3ff904dbb5fef896":{"kind":"REWORK","_number":1,"created":"2022-03-31 13:16:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/19/836019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/19/836019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/1"}}},"commit":{"parents":[{"commit":"1dd62c1854bc08d2aee17845b23b93084520d46c","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1dd62c1854bc08d2aee17845b23b93084520d46c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-31 13:16:01.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-31 13:16:01.000000000","tz":0},"subject":"Add allow_ping_health_monitors parameter to api","message":"Add allow_ping_health_monitors parameter to api\n\nThis can be used to set the allow_ping_health_monitors\nconfiguration option in api_settings group\n\nChange-Id: Icdee753af276224c969a3eb0e790d331269893d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae2f83a3a54dd17bd45831ad3ff904dbb5fef896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/ae2f83a3a54dd17bd45831ad3ff904dbb5fef896"}]},"branch":"refs/heads/master"},"c3d78e5cb45d0a5d5c360746c2c06298b638ec57":{"kind":"REWORK","_number":2,"created":"2022-04-01 01:04:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/836019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/19/836019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/19/836019/2"}}},"commit":{"parents":[{"commit":"1dd62c1854bc08d2aee17845b23b93084520d46c","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/1dd62c1854bc08d2aee17845b23b93084520d46c"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-03-31 13:16:01.000000000","tz":0},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-04-01 01:03:45.000000000","tz":540},"subject":"Add allow_ping_health_monitors parameter to api","message":"Add allow_ping_health_monitors parameter to api\n\nThis can be used to set the allow_ping_health_monitors\nconfiguration option in api_settings group\n\nChange-Id: Icdee753af276224c969a3eb0e790d331269893d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3d78e5cb45d0a5d5c360746c2c06298b638ec57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c3d78e5cb45d0a5d5c360746c2c06298b638ec57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
