)]}'
{"id":"openstack%2Fpuppet-octavia~842079","triplet_id":"openstack%2Fpuppet-octavia~master~I9a2cfddfa2e91fde8280c4cab4c51d05ae436ae5","project":"openstack/puppet-octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-05-19 13:04:26.000000000","reason":"Takashi Kajinami replied on the change"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2022-05-23 19:32:51.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2022-05-23 19:32:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a2cfddfa2e91fde8280c4cab4c51d05ae436ae5","subject":"Allow amphora timezone configuration","status":"MERGED","created":"2022-05-17 10:32:31.000000000","updated":"2022-05-23 19:34:03.000000000","submitted":"2022-05-23 19:32:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842079","meta_rev_id":"9d14797b047f6c353f49c961e974479bd83be689","_number":842079,"virtual_id_number":842079,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-23 19:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-05-19 13:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-05-19 12:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-05-19 13:04:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 10:37:59.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2022-05-17 13:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 00:33:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-05-19 12:25:18.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ff60540a00f7151095a9cc2cec8735179ba9c829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-17 10:32:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"003c6a49e637b50b4ddb9ebdadaf1968d026395d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-05-17 10:37:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af68c011f1b6383659a536ce8a635b1bddc5688c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-17 10:45:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6183b1e41df4ddfaadd0e3564d7e83c4c11c443","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-17 10:45:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec9a0d18116bf8f101bf23a0ce4eacaa6103d344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 13:07:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e31b8ebb5284383a678ddd8bcc2c72d : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/436af59dd5b949a6954bbd81e0781d16 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81bea32e95f143869dba10e20c66a2a8 : SUCCESS in 3m 43s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/9eb61aa3e4754bce90ba3b90e2f0afbe : SUCCESS in 11m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace58f82caae42758f0f9e828c3089ec : SUCCESS in 8m 59s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/997fa1b6c9874f8cbec016ca868bd356 : SUCCESS in 9m 11s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c64395bf64b4b259613413316c34945 : SUCCESS in 8m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0734c0077b984f7aa4fd2fd96ab099ea : FAILURE in 23m 50s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/193c70fb2d264babac8418d535ede1f1 : SUCCESS in 47m 26s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/621cffd10d1046998579313c74b4b661 : SUCCESS in 19m 54s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/731a86f2ab974b408088b5a6d7886f4e : FAILURE in 38m 01s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8365064b04a4634aeb8caa33131a29d : FAILURE in 11m 01s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c172f3dfbf8045378c55c91e9e2cc384 : SUCCESS in 19m 45s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aba98f1732e4bafac704cc4210e967b : SUCCESS in 20m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d207fa086a1744a08ddfcbfa8613097a : SUCCESS in 4m 58s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f05db407e7554a45b888b8ea76a7827d : SUCCESS in 2h 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72f4ee83975c96a6237ba9be7f614a3f0db52deb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-17 15:03:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"973a5781233190ccc8cf38a5f101d544da2de6b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 17:14:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b98f2b1a7b84155a7a695bcae9a8744 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3e2809b3d524773a5adb67228534dc8 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5441d016fd84530a28e3b98eb68561c : SUCCESS in 3m 30s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac56dcb860a74491a68b141c0d423c0c : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52601b7b653f437086ea2caf9e590b70 : SUCCESS in 10m 29s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8268daec23914d8d9547c985461b1bbf : SUCCESS in 10m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/600441499bbb429a8942622531d46cd3 : SUCCESS in 11m 27s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dbccc63b0444e4db77212137401b92d : FAILURE in 23m 51s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a61b75cec4004723984c43bffbd559e3 : SUCCESS in 57m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4590db3ff6fe4bdab5458936b0ff62d9 : SUCCESS in 20m 18s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/76dd767de1eb41788e8871550dc1a96c : SUCCESS in 30m 34s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/20595017b8c046078db4d1b08d1e773c : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8842460ff1464a31b0604690ec1b3ecd : SUCCESS in 19m 37s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70842424976a44b0858b1e15f2875402 : SUCCESS in 19m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/176a43898a0845a98719d446e01d7551 : SUCCESS in 3m 45s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b5ecf29987444aa89fe3506de7fec133 : SUCCESS in 1h 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"34b5f501777d5faa33463f7f1db270665339de24","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-18 00:33:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"13b3e5cee37ff875082db8d52296bbf6a0dfd09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-19 09:52:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"941a56e6f335546002135ae88991b13b758bc019","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-05-19 09:54:18.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"339d41edc5b8eee1ade8ca4c30d23dcdc241612b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 12:24:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60407f28617e497fa10219abf45d96a1 : SUCCESS in 3m 09s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c0bdd732614871b97a98cefd36e75d : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e430124d0b4f40dbb85d0a59c6ae2d89 : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/f98ab50c019d42d582fc7dcc2574e4b6 : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/370d5c7354ad4d6bb580cc8600a1cc60 : SUCCESS in 11m 00s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22e728d9cd444bd2aea746194e38df0f : SUCCESS in 10m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb3b4a614dc490f89479467e5117245 : SUCCESS in 10m 06s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77733ed24cb443988d0c5e5570f17cd6 : FAILURE in 24m 34s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/534c5b339133441fbca392f8234e08b5 : SUCCESS in 56m 50s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d383e95a392d47b98c0c50fddf70b877 : SUCCESS in 22m 04s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/2373f6297c324838adbed530d457cd3f : SUCCESS in 31m 04s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/318161108e194023900feed42da91900 : FAILURE in 9m 14s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd6432c009b34d9cb2b0278e704999db : SUCCESS in 20m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b46d792d417430f9adad1706b2d3a04 : SUCCESS in 20m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0a354d60c474002a5dcab33d78a5985 : SUCCESS in 3m 17s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b1c4cc73545b46398fde148a45d3c281 : SUCCESS in 1h 59m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2911292bddd17e7f8f69e65fb2f1cbaebfbc5de6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-05-19 12:25:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a595df4543700c45ef6dd5db87b1ef83c7ebbcca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-19 13:04:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3fbeabbf99b3397bf951ab40ecf4d4bf204b320b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 13:06:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e0e20bc2c5e7164a89f32875d4e92bef2ce077e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 14:00:55.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c20fdee63f490a855af7d3f9593023 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/866219c37ab242bfa897ae7c1908e559 : SUCCESS in 3m 07s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79c440f0ee9e4e52a697ed9983fd6f98 : SUCCESS in 3m 47s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e3bf223bd804ce9808a5438f8abd5e3 : SUCCESS in 11m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0651b49d366f4c45ad41cadc782a05ff : SUCCESS in 9m 04s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b081ac39354e35b3f2fb324a678b6a : SUCCESS in 11m 49s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5dc3353258a407a8ba516623a646a4a : SUCCESS in 50m 23s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2270d5bc7294fe58d9d63dd9936a2f8 : FAILURE in 32m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2075a78b337e41b59bdda066d0854eb0 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"9a81898b0943decb3e525dc8a3358f7c438e9125","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-22 14:43:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d65acaf48837004836bef810c7f2b92ed1ca427b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 16:31:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ccafa9d3c040cba77fa26b524e57f4 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c10187d6b25d4284bce99e6cbb95275a : SUCCESS in 4m 37s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45ca522b49dc43a8bbc71d6a2fc8f58e : SUCCESS in 2m 41s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5dd0dd2099fd45a69cf09cb6c8db6db6 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/428294b6406f40deba72917ddd52a3c0 : SUCCESS in 11m 28s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aa27044703f444485fdecb94df223fb : SUCCESS in 12m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32c30b3aeeaf4ebab40694edadea68bf : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0adf26c5ae7a4cc1884c2ce1421f1350 : FAILURE in 25m 47s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf321226237444dbbb9c33aa4f6ac917 : SUCCESS in 47m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/940b7990dae943649951282d75582183 : SUCCESS in 21m 22s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/1bf7048d571948e5a0cd1e15bac2cd26 : FAILURE in 16m 10s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c0ec331ea394b63bfba6ce7fb561425 : FAILURE in 8m 21s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d425d98c75e0417e9703bc81b966624c : SUCCESS in 19m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4209f707d1a74e419063ec6ef3f5d8be : SUCCESS in 17m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86ffd7b115ea4345ae13012d5b82016f : SUCCESS in 3m 05s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3752c9a4a13f43c69fa7da8141102462 : SUCCESS in 1h 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8d57d3d457b6b6e9de3ea0e198bebf8807af0fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-23 16:34:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8641f9bdf8b5403ec27030c2aa06bfe916431b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 18:43:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18a5e372a49749eaa3534c4ebbc0cf7c : SUCCESS in 2m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9192dc803e654a52b52dc6e6888f6981 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f0d4190e6a8494990977048ec276f94 : SUCCESS in 2m 53s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/5369f82d9441458db3351d5c51336630 : SUCCESS in 12m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d0f8d0666f94d58ad3989fd65e3e94f : SUCCESS in 11m 27s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f90e0dd88eb3481492dc939540a945eb : SUCCESS in 9m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0d9eedb5ebb4b48ab8b6c2a6f824fde : SUCCESS in 10m 17s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e113e73366f646438dd67521d947e10a : FAILURE in 24m 23s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b0469bbc9c54b17b0519a0dacf77517 : SUCCESS in 47m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac4f11de5d34476eb44a292804b6826b : SUCCESS in 20m 14s (non-voting)\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/fef24174e4f64259b204f543f1b5404a : SUCCESS in 23m 46s\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/975776138add4d1bab84f65caca69ebf : FAILURE in 9m 42s (non-voting)\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46b9d1a8c7234604a5492c593220d897 : SUCCESS in 21m 54s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37b55a841c3345f9a90224ae585dd8d9 : SUCCESS in 18m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ef6e0b5bb18416c9e8d4411e2f877b6 : SUCCESS in 3m 16s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/651be8d6ba0541a494052e08624a408b : SUCCESS in 2h 06m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"084bbf67883027bf27a7aaba6e2fc763c442c33b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 18:44:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5f46c79a0465fef4080e712894d93c93e97bdb01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 19:32:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/263da27f6abe41c69ba90bea679dde2c : SUCCESS in 3m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7df696959ae4430a8eeb93aa42ef1b70 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac0c605b93f049b4a69efd5443671e8e : SUCCESS in 3m 56s\n- puppet-openstack-unit-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/4526ddae3d5840d5be468029f4263352 : SUCCESS in 13m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f6057535c064255a7e2e3c303d34883 : SUCCESS in 8m 29s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8615b3f2b1c54c7cb317697447b18368 : SUCCESS in 8m 25s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/030bb07f6e804e898dde37c5d0e3f461 : SUCCESS in 43m 37s\n- puppet-openstack-litmus-7.10-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec34b86156f34f1bbc8123cc49ae21a7 : SUCCESS in 26m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d763f07add264de4bbb52ac12b32337a : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"24c51c6d071f05693d97a7cffa0c2714f748c14c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 19:32:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9d14797b047f6c353f49c961e974479bd83be689","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 19:34:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a91022fa1a745aebc2b0218bf476a17 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b5ab3de4bcf122d4228173eca0d0cd733a0e1ed9","revisions":{"cc6d6dfe2537f441928ac40102f56570d5d22c2b":{"kind":"REWORK","_number":1,"created":"2022-05-17 10:32:31.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/79/842079/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/842079/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/1"}}},"commit":{"parents":[{"commit":"c7dd10a34f204641c0f47d1d912a509f555c6a1a","subject":"Replace remaining hard-code of OS user/group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c7dd10a34f204641c0f47d1d912a509f555c6a1a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-17 10:30:18.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-17 10:32:29.000000000","tz":120},"subject":"Allow amphora timezone configuration","message":"Allow amphora timezone configuration\n\nAdds support for the controller_worker/amp_timezone configuration parameter.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/839937\nChange-Id: I9a2cfddfa2e91fde8280c4cab4c51d05ae436ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cc6d6dfe2537f441928ac40102f56570d5d22c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/cc6d6dfe2537f441928ac40102f56570d5d22c2b"}]},"branch":"refs/heads/master"},"c8a483d1862f2ffc43a14e5bbe50f96fb9ab998c":{"kind":"REWORK","_number":2,"created":"2022-05-17 10:45:38.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/79/842079/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/842079/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/2"}}},"commit":{"parents":[{"commit":"c7dd10a34f204641c0f47d1d912a509f555c6a1a","subject":"Replace remaining hard-code of OS user/group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c7dd10a34f204641c0f47d1d912a509f555c6a1a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-17 10:30:18.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-17 10:45:23.000000000","tz":120},"subject":"Allow amphora timezone configuration","message":"Allow amphora timezone configuration\n\nAdds support for the controller_worker/amp_timezone configuration parameter.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/839937\nChange-Id: I9a2cfddfa2e91fde8280c4cab4c51d05ae436ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c8a483d1862f2ffc43a14e5bbe50f96fb9ab998c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c8a483d1862f2ffc43a14e5bbe50f96fb9ab998c"}]},"branch":"refs/heads/master"},"b5ab3de4bcf122d4228173eca0d0cd733a0e1ed9":{"kind":"REWORK","_number":3,"created":"2022-05-19 09:52:41.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/79/842079/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/79/842079/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/79/842079/3"}}},"commit":{"parents":[{"commit":"c7dd10a34f204641c0f47d1d912a509f555c6a1a","subject":"Replace remaining hard-code of OS user/group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/c7dd10a34f204641c0f47d1d912a509f555c6a1a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-17 10:30:18.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-05-19 09:52:01.000000000","tz":120},"subject":"Allow amphora timezone configuration","message":"Allow amphora timezone configuration\n\nAdds support for the ``[controller_worker] amp_timezone`` configuration\nparameter. It allows to configure the timezone of the amphora.\n\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/839937\nChange-Id: I9a2cfddfa2e91fde8280c4cab4c51d05ae436ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b5ab3de4bcf122d4228173eca0d0cd733a0e1ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/b5ab3de4bcf122d4228173eca0d0cd733a0e1ed9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
