)]}'
{"id":"openstack%2Fpuppet-octavia~847651","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"provider_drivers-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-06-28 11:24:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2","subject":"Accept array or hash for enable_provider_drivers","status":"MERGED","created":"2022-06-25 12:21:54.000000000","updated":"2022-06-28 11:27:03.000000000","submitted":"2022-06-28 11:24:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847651-provider_drivers-stable/wallaby","cherry_pick_of_change":847650,"cherry_pick_of_patch_set":1,"meta_rev_id":"2a27754b963e4b624bb3c3f9a23856a2d42bc39c","_number":847651,"virtual_id_number":847651,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 11:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-06-27 15:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-06-28 10:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-06-28 10:48:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-25 14:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 15:29:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-06-28 10:48:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"30c495cea9d5fde563ab696b0026d1c0bfb9aaae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-25 12:21:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"2e39012fbecfb322bd3c9d2fa2f4cbd475128d2b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-25 12:22:02.000000000","message":"Topic provider_drivers-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"cb09617c3ca54f5aa3899d3a553f8bbde43e06e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-25 12:22:07.000000000","message":"Topic set to provider_drivers-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"d500067f2a32d96d263153293f449b9fbb2ab428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 14:27:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac8b4df04b8a41ce9ec32c8488feab5c : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bab52d77decf4c6494a12f0b99644b4a : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff21aba616cf4b2881e0462be63ff1de : SUCCESS in 9m 52s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d725e2bd405b40fa8fedfbc84aaefec3 : SUCCESS in 11m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d7f95a7c212406ca6ed9fe18f7cecf9 : SUCCESS in 7m 09s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4ee6c8ccf9a4921960d8cd7112862e9 : SUCCESS in 27m 22s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4e268c569284fd5979386cb3dfbd9d0 : FAILURE in 10m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6eef4740b26452aa3d9ef0b9c4e9e2c : SUCCESS in 4m 08s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c3de6de7c2b4448ca83fbfaea44ec6e8 : SUCCESS in 1h 49m 33s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fb56bf698e4d4c4ebefdc1b1988bb84b : SUCCESS in 2h 02m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2047dc9e6bf44dc22a905d86f8e7dc161663abd3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-06-27 15:29:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa59b1094700737e01aebbe490c3079694744023","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 10:48:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20e0fa4ede88b1239cba103bc96912e9eefec973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 10:49:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73b9ccc94e212640b12e2361daf59c7612143702","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:24:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a4889a50c2241018a5d22cb7dc6862f : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f2ff7066f8e4d55b78caf1543fcce1e : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/526fd7875b53404a951fc0c7ea701e82 : SUCCESS in 13m 02s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1239477735548498bd81050da412b18 : SUCCESS in 10m 54s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/870c16c4fa094a6ebf19dc4b9f5b1206 : SUCCESS in 29m 40s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95db714078ff44408dead86fefd6576a : FAILURE in 14m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb57aeae5df74db785b9db9fcd30d6ac : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":1},{"id":"194d416272dc3ff6175d98a88bd20444b4f4eed6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:24:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2a27754b963e4b624bb3c3f9a23856a2d42bc39c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:27:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c650cbe4fa7848adb0c1ac28db542251 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71e762c68f0b98ae37969f13497aa0a364889255","revisions":{"71e762c68f0b98ae37969f13497aa0a364889255":{"kind":"REWORK","_number":1,"created":"2022-06-25 12:21:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/847651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/51/847651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/847651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/847651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/51/847651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/51/847651/1"}}},"commit":{"parents":[{"commit":"765afdf3a9e22172674b6dc1e43de648d914a27d","subject":"Add parameters to configure amphorav2 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/765afdf3a9e22172674b6dc1e43de648d914a27d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-02-23 09:51:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-25 12:21:54.000000000","tz":0},"subject":"Accept array or hash for enable_provider_drivers","message":"Accept array or hash for enable_provider_drivers\n\nThe [api_settings] enable_provider_drivers parameter is DictOpt and\naccepts a string value which represents a dictionary. This change\nallows using array or hash for this parameter so that users can use\na more \"native\" type to define its value.\n\nChange-Id: Icfa01a6cba5f4eda68d51ec2740f24fbc049f6e2\n(cherry picked from commit 270d33d5436b0e9f13f88b308a17e299e00f21bc)\n(cherry picked from commit 8da4abe08def306e2f3669f5aa2dbfb8d1d69459)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/71e762c68f0b98ae37969f13497aa0a364889255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/71e762c68f0b98ae37969f13497aa0a364889255"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
