)]}'
{"id":"openstack%2Fpuppet-octavia~848370","triplet_id":"openstack%2Fpuppet-octavia~master~Ia75fb9a1b343293c2b962ae4f90894e094149acc","project":"openstack/puppet-octavia","branch":"master","topic":"apache-headers","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-30 13:55:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia75fb9a1b343293c2b962ae4f90894e094149acc","subject":"Expose headers option of apache::vhost","status":"MERGED","created":"2022-07-01 01:07:43.000000000","updated":"2022-07-30 14:56:46.000000000","submitted":"2022-07-30 14:54:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848370-apache-headers","meta_rev_id":"802bb4f488270af285b7f5237c0bd7b4d4fe2851","_number":848370,"virtual_id_number":848370,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-30 14:54:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-07-30 13:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-07-06 14:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-07-30 13:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 03:42:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-06 14:23:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"aab8be34fcd0ecba1205d822bad8752df8f9972d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 01:07:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d74ba88b0c3e56ffbd2ab4058c4521a747941d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 01:26:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85e9e0d4fe9534b32bbf3e8260b5d774ae419eb9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 01:28:49.000000000","message":"Topic apache-vhost removed","accounts_in_message":[],"_revision_number":2},{"id":"32186b10bffc3083459da0edf498c98ed96979ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-01 01:28:51.000000000","message":"Topic set to apache-headers","accounts_in_message":[],"_revision_number":2},{"id":"244afbed54e089f1419932169b0c8554c51971fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 03:42:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b64a7ce54a3446aaa4c5480f66e3e735 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d31ea86b44473c931fc14bbd6b1d55 : SUCCESS in 2m 59s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/091027d6bf1a4d488a8c5158efcf829d : SUCCESS in 2m 59s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cfb8fe741134ce8b2dded01b752fee5 : SUCCESS in 11m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/258d17d91f2345dfbf392d774ffecd39 : SUCCESS in 8m 43s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f7019c4fa624191ae31e874c00312cd : SUCCESS in 9m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/007c679abf9f416fb4ed3a84c83fc3c5 : SUCCESS in 8m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1630bb5e62d446b2ba986aac8c62e420 : FAILURE in 23m 10s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c272af4ee144e4587a069be67b04f0b : SUCCESS in 56m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6109ed6d2b104c85b380c02dd582af6f : SUCCESS in 20m 27s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/377937bebd8840b892c263e13da9505a : SUCCESS in 24m 41s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/809a144b1b3a4bffab5b9ef1ea3b9cd1 : SUCCESS in 20m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67ad5b1e4add43d5bf46fc04e92c22b4 : SUCCESS in 20m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0778fe9139b7453eb047d1683cf47c29 : SUCCESS in 3m 20s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c222fe3c83a148799a438804e9a1632a : SUCCESS in 2h 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6f3cb1e70ab188d39e0d31557601fc863478955","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-06 14:23:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"413849dc340480aed7f6ae144a02f514d522e9aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-30 13:55:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"523236a68430a779603be503976792b7bf0fd800","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 13:56:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8446bf9f5d1b9475db7762ec971fb4c97a6dd1f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:54:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f8ad5330f804f3bb7eae15441cf8538\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ceb54fe8174607a66068934d421e81 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddac95c1020b49dd91c0a975a058ddd5 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb04563ee4ad421286ddd26de4e47415 : SUCCESS in 3m 44s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f332d99fc19e4556ae4ee01d5280664e : SUCCESS in 15m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/686f0283639744b2a3a1fe74a1ca2abc : SUCCESS in 10m 08s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/747b9db4fd8e412ca98b63771f86bff6 : SUCCESS in 11m 22s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6ecff22efb742a7bbb32a7d2b27ff4f : SUCCESS in 55m 12s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/90bfb8fdd11e452194c5c2520cabbed2 : SUCCESS in 26m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed611ebe86624e3d9f893987164ae5fa : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d943a9855f29550fc4c012d8077de345e89ec456","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:54:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"802bb4f488270af285b7f5237c0bd7b4d4fe2851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 14:56:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f81c6d4968474598928f1e395bf55f9a\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6095a67beb63467b8d05b9306dd3c7f8 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d7dc6e1a4f256686484c83b97752f17b5ff04bf1","revisions":{"83924bd149b78c0413ba1da649ba4fa2b6a95e24":{"kind":"REWORK","_number":1,"created":"2022-07-01 01:07:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/848370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/70/848370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/1"}}},"commit":{"parents":[{"commit":"76851562562f047f0efcc565317797176e1a614f","subject":"Merge \"ovn: Accept arrays for ovn_*_connection parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76851562562f047f0efcc565317797176e1a614f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 01:07:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 01:07:34.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: Ia75fb9a1b343293c2b962ae4f90894e094149acc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/83924bd149b78c0413ba1da649ba4fa2b6a95e24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/83924bd149b78c0413ba1da649ba4fa2b6a95e24"}]},"branch":"refs/heads/master"},"d7dc6e1a4f256686484c83b97752f17b5ff04bf1":{"kind":"REWORK","_number":2,"created":"2022-07-01 01:26:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/848370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/70/848370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/70/848370/2"}}},"commit":{"parents":[{"commit":"76851562562f047f0efcc565317797176e1a614f","subject":"Merge \"ovn: Accept arrays for ovn_*_connection parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76851562562f047f0efcc565317797176e1a614f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 01:07:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-01 01:26:40.000000000","tz":540},"subject":"Expose headers option of apache::vhost","message":"Expose headers option of apache::vhost\n\nThe headers option in apache::vhost is required in some case, for\nexample when adding the X-XSS-Protection header. This change allows\ncustomizing the option for the api vhost.\n\nThis change also adds support for request_headers so that both request\nheaders and response headers can customized.\n\nChange-Id: Ia75fb9a1b343293c2b962ae4f90894e094149acc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d7dc6e1a4f256686484c83b97752f17b5ff04bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/d7dc6e1a4f256686484c83b97752f17b5ff04bf1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
