)]}'
{"id":"openstack%2Fpuppet-octavia~849009","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~I9542ca07f8fd7da442ce3d6aab0f88286db021ad","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"ovn-tunables-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-19 12:52:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9542ca07f8fd7da442ce3d6aab0f88286db021ad","subject":"ovn: Add more tunables related to OVN DB connection","status":"MERGED","created":"2022-07-09 15:05:26.000000000","updated":"2022-07-19 12:54:06.000000000","submitted":"2022-07-19 12:52:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849009-ovn-tunables-stable/wallaby","cherry_pick_of_change":847269,"cherry_pick_of_patch_set":1,"meta_rev_id":"d26a66e6f91db45afd20b8a107ad7d899feb3696","_number":849009,"virtual_id_number":849009,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 12:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2022-07-18 15:45:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2022-07-11 22:00:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-18 15:18:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2022-07-19 12:20:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 17:31:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 22:00:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-18 15:18:54.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-07-18 15:45:20.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2022-07-19 12:20:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"373eb7ef2186fca4472a7d193ca4728a0817af54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:05:26.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"a0310de2e8a163ab78ae2ec114279287fdcccc11","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:05:33.000000000","message":"Topic ovn-tunables-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"85ba7aea5331e27f21e054b1a2069daa4623c67a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:05:37.000000000","message":"Topic set to ovn-tunables-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"a6b9c6a309767d13724112211d7d0a1d647e8822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 17:31:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0d9369019e54c86825131f7332f4cce\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30d6d7baa4424a43bb4137f24521e140 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/438cb6a138634fd198ee1e1dd7831e85 : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/57de25ed4073467e91fff965501a6912 : SUCCESS in 9m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00098f8b981b4b60811d9ebefc5329e0 : SUCCESS in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e21a4c3ec4b467e84d09e0937d2d407 : SUCCESS in 6m 45s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bdb9f8166754cb8b3ca0b17bbf2694b : SUCCESS in 27m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5a9f9958f294f7f9fc36daab4719f26 : FAILURE in 10m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89aac20499be4fc5ba9c658f01e9e0d0 : SUCCESS in 3m 07s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4575a0153d4a4535a706febc06f6dddf : SUCCESS in 2h 18m 10s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/67c400ed77764370ad0363db5f7742c1 : SUCCESS in 2h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfb89b1e54d71535950cc14f3c5160a5a07422a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 14:55:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8096dd6753cec6d0de2e07d95651b1366a73a3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 17:10:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3239a2f485404616bf89a553a3095319\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af3b79dc83e84eed9636630fd8d826f6 : FAILURE in 6m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/000384f30fa3492781c842f41c401392 : SUCCESS in 6m 22s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0154a3c689d14c23b3b3a0a68dbb0cf0 : FAILURE in 15m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b7b6048fee94660bae98ed4bc14b23a : FAILURE in 19m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e42b668dfb499896f1f029dfa0446f : FAILURE in 11m 07s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/13272f1f182549efb5199cce6763b328 : SUCCESS in 42m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af265401040d4daab10456a9082b2f85 : FAILURE in 15m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7b9ff3fc976480187786aa40bbc7e74 : SUCCESS in 3m 59s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/554e06dc2d3a4216a5c6960f602a34db : SUCCESS in 2h 07m 32s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/c7fb0d48241d405cb618524cd1669902 : SUCCESS in 2h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8791317713e3c97c019d94e03f4f49c92443966d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 18:12:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"69e11e33c04b9d530544a96cb9e0edc4c9bc6d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 20:55:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef68aa113594317ba09b50d2ee39528\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/811fce4f4eb74447af3e16278ed257a9 : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef4fce566fb6486b88ab5a45e0cef04f : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7cfce12f23f141d8af04c0b2699f3f52 : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/075479d8f599441ab805b06dc4ff075a : SUCCESS in 10m 33s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc2fe742ecb4b2fa64de4120c9bf378 : SUCCESS in 10m 37s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/084b8afba711461287af17cba494fc1e : SUCCESS in 25m 45s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a9c838d9f7a43ffa1738a5788746342 : FAILURE in 15m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39b592ce3d3b4516aaadd0213cf5fab2 : SUCCESS in 3m 49s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2f8d9467f45d41f899cf7cb1756bbe6a : SUCCESS in 2h 39m 14s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4fc44bb991c44230ac57fa56d8d4cb54 : SUCCESS in 2h 07m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"145b3cbcf528e3bb4ee62966d850a237d76a6ee5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 22:00:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"92bb752ec788a9c0b71d8feef2272c56fc685fa4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-07-18 15:18:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"526f6b098be40fb0983255886e84fc98a9905f85","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-07-18 15:45:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8a0fb3c67cee627cbb0b00921c86d1ca0021a6e5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-07-19 12:20:41.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"279703ddc7c57ac45ac96d502830aae40ac2f5d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:21:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb5b04635f7af8b0132ae9fdc0bbbfc2631c1c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:52:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80d01210286409bba7d97f8365897fc\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ef605d938da4ca9b8ffa9d78158b0e8 : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55db3b845a7648d98c0dc776cab7cfdc : SUCCESS in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7d447dfa8d4482cb33a5348eb7834e4 : SUCCESS in 23m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba3eb31353ce45faabf8b09ed396fe02 : SUCCESS in 11m 04s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/adbb3da403994455af9543f682869d69 : SUCCESS in 27m 04s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/387dd401d555412c8ff1749db565e664 : FAILURE in 14m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f505ac1b99d04539ac7e02df2f3d9538 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6afe0bd1cbf69d89e46d8206d2ff1d0bf57384a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:52:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d26a66e6f91db45afd20b8a107ad7d899feb3696","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:54:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d33113b1fe4acab4bf612db67e0649\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd12188fa88647c7814178690de28aab : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0c8f1c4dbf843ef38bba03ec08957840fa05393c","revisions":{"bde5ae021dd874ee50fae1da2bed4fcd19219d44":{"kind":"REWORK","_number":1,"created":"2022-07-09 15:05:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/849009/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/09/849009/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/1"}}},"commit":{"parents":[{"commit":"be4b1eda16119e211a146175b1771c4c74718913","subject":"Manage ovn-octavia-provider package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/be4b1eda16119e211a146175b1771c4c74718913"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-09 15:32:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:05:26.000000000","tz":0},"subject":"ovn: Add more tunables related to OVN DB connection","message":"ovn: Add more tunables related to OVN DB connection\n\nChange-Id: I9542ca07f8fd7da442ce3d6aab0f88286db021ad\n(cherry picked from commit 2a6d6b8b4d124b16536c782030b768d9b0294c5e)\n(cherry picked from commit d64492c34a90b724e521cb3a36baa2f9652232ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bde5ae021dd874ee50fae1da2bed4fcd19219d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bde5ae021dd874ee50fae1da2bed4fcd19219d44"}]},"branch":"refs/heads/stable/wallaby"},"a9186d19cf5262d7e59447e3c0e28c664fece677":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-11 14:55:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/849009/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/09/849009/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/2"}}},"commit":{"parents":[{"commit":"26bb0452cd3f5f10f8c4226400d4f78214c6275f","subject":"Manage ovn-octavia-provider package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/26bb0452cd3f5f10f8c4226400d4f78214c6275f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-09 15:32:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 14:55:15.000000000","tz":0},"subject":"ovn: Add more tunables related to OVN DB connection","message":"ovn: Add more tunables related to OVN DB connection\n\nChange-Id: I9542ca07f8fd7da442ce3d6aab0f88286db021ad\n(cherry picked from commit 2a6d6b8b4d124b16536c782030b768d9b0294c5e)\n(cherry picked from commit d64492c34a90b724e521cb3a36baa2f9652232ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a9186d19cf5262d7e59447e3c0e28c664fece677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/a9186d19cf5262d7e59447e3c0e28c664fece677"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"0c8f1c4dbf843ef38bba03ec08957840fa05393c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-11 18:12:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/849009/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/09/849009/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/09/849009/3"}}},"commit":{"parents":[{"commit":"f2bb105bbd48d1b876f1b4defea9e0a99dc495d4","subject":"Manage ovn-octavia-provider package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/f2bb105bbd48d1b876f1b4defea9e0a99dc495d4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-09 15:32:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 18:12:46.000000000","tz":0},"subject":"ovn: Add more tunables related to OVN DB connection","message":"ovn: Add more tunables related to OVN DB connection\n\nChange-Id: I9542ca07f8fd7da442ce3d6aab0f88286db021ad\n(cherry picked from commit 2a6d6b8b4d124b16536c782030b768d9b0294c5e)\n(cherry picked from commit d64492c34a90b724e521cb3a36baa2f9652232ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0c8f1c4dbf843ef38bba03ec08957840fa05393c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/0c8f1c4dbf843ef38bba03ec08957840fa05393c"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}}]}],"submit_requirements":[]}
