)]}'
{"id":"openstack%2Fpuppet-octavia~849211","triplet_id":"openstack%2Fpuppet-octavia~stable%2Fwallaby~I725d7c167886ae799663bbe3c033fb99a81f9711","project":"openstack/puppet-octavia","branch":"stable/wallaby","topic":"ovn-conns-list","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2022-07-18 15:43:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34451\u003e replied on the change","reason_account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-19 12:54:38.000000000","reason":"Change was submitted"},"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2022-07-19 12:54:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I725d7c167886ae799663bbe3c033fb99a81f9711","subject":"ovn: Accept arrays for ovn_*_connection parameters","status":"MERGED","created":"2022-07-09 15:06:31.000000000","updated":"2022-07-19 12:54:38.000000000","submitted":"2022-07-19 12:54:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849211-ovn-conns-list","cherry_pick_of_change":847889,"cherry_pick_of_patch_set":1,"meta_rev_id":"b2d78c3a485777344293132caaf57e5eb4bd94c7","_number":849211,"virtual_id_number":849211,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-19 12:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2022-07-18 15:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":2,"date":"2022-07-11 22:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-18 15:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2022-07-19 12:20:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-09 17:32:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 22:00:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-07-18 15:20:48.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-07-18 15:21:04.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2022-07-18 15:21:04.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2022-07-19 12:20:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"}],"messages":[{"id":"e7e1a4e6d352e4fa11456f73ea1f498ecee82b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:06:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"cba67eade386db58344a24c0abbba3fc5cc182d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:06:46.000000000","message":"Topic ovn-conns-list-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"84e68a5ece2732441197a8e361c63fa6fa4898da","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-09 15:06:50.000000000","message":"Topic set to ovn-conns-list","accounts_in_message":[],"_revision_number":1},{"id":"4e583a2da26584d6d8726dc553b32550fc5f4f42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-09 17:32:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f776c0ac7a44feb82ad3b93ac25c07a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e1cb4ac4974c4186adfb849052b87f : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77260874fbb9410ea925529104bbfc7a : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a872bfb550ac4e4ca518f72dd9686961 : SUCCESS in 11m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00312e4f0e9444fb4246d60d661cba7 : SUCCESS in 8m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c27b4044f2b4c5d82639864182c73e0 : SUCCESS in 8m 12s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/acffbac827df4d41b72712b643d7e589 : SUCCESS in 28m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22936b751f4c4aae865bc824118ebc80 : FAILURE in 9m 35s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6bac031c8a0948dba70772322483dcff : SUCCESS in 2h 19m 10s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f7de8ebac0e041dda4b8be367207892e : SUCCESS in 2h 13m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65709a1838659552bb7bcea21f127d62089c9764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 14:55:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e16fc738d021a49e4f2d5a192445dad508ff10e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 16:58:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8aa765d43158432ca47fe5aa7b5ba3cd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/937bb4851a87433e8bd47ee3fd075cea : FAILURE in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/862f45004a304cbc837df1865eb40274 : SUCCESS in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1bc632333e24118b1626f547487d87b : FAILURE in 13m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb960e666e6e42e8adc37dbe202c83aa : FAILURE in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a436d296da24bad9ebae00719e87cab : FAILURE in 8m 00s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a827308904ef43a1b295e9f154c2a540 : SUCCESS in 29m 56s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75164b70ab3f41cf84013f2beb35dd09 : FAILURE in 13m 04s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/62e1b6d748304ddf8e9b1a646cb7e36f : SUCCESS in 2h 00m 30s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/66460b009cbc433798060e8d091d4d4b : SUCCESS in 1h 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d96583a838ab12363129818622fe2a7b52db00eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 18:13:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5dac139b5e71cbcc8768b90644fe9b598069b7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 20:46:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a508c0f0d92d4cd38e06f17844f35de8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16f554f92f9242188d81e887e5faa0f5 : SUCCESS in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/014194234f734c9a9a6afe6fdc21339f : SUCCESS in 2m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/df1d9c758d054f178e423b263ecca53d : SUCCESS in 12m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/427937c428b0400b9281a93410331bf5 : SUCCESS in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dbf1499a55746daa5ebdeec654a32e1 : SUCCESS in 7m 34s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/fceeeaecb6f44b56807411d139b152d1 : SUCCESS in 27m 10s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac123b1501d94bc9ad78263c3125fa83 : FAILURE in 12m 04s (non-voting)\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9775350b796f4049aa1e719a420afe06 : SUCCESS in 2h 27m 01s (non-voting)\n- puppet-octavia-tripleo-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8bacad8da6294a90a10ff31a1f3f84a2 : SUCCESS in 2h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ba45ff0fd82e743003cab5d8642dbd5b6d9a34c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 22:00:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"024ff42434707cbae9ff36850780e1fd1a6befc0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2022-07-18 15:20:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f3a411e8c2ea6691b86006813de24c56a8f9e0bb","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2022-07-18 15:43:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a197eb56c8e5e01e65e8ff89c62a89fc7cc8de5c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2022-07-19 12:20:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5b5bda845335fdccf5539b08796d6bd4aa25e09c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:21:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a2fe0f6e0ca93f7e017d5f6fd32f031341bedaf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:54:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d12d5979c6274dd294102c69c14f4b5a\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10012f287b524fc79511580815f6bfdd : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0831032bd1a9480abe9aeaf226720bdd : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ddada45a031f4f6997f6a3eb1174bd1f : SUCCESS in 24m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/056244ab88cd4ab381772d56764dfd8d : SUCCESS in 12m 42s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f76efd00b5354f01a59abb66e11c5a84 : SUCCESS in 26m 55s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1bdd1d58524e24ba159b209994ae54 : FAILURE in 17m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2d78c3a485777344293132caaf57e5eb4bd94c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:54:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b2e58063ff6f1cbb2e470dc6ebd1a60482eaa43","revisions":{"bab76fd047508841f4bddc4f2350b98d0896860d":{"kind":"REWORK","_number":1,"created":"2022-07-09 15:06:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/849211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/11/849211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/1"}}},"commit":{"parents":[{"commit":"30aca97f35d22ad842a39a4625de792b1a07a246","subject":"ovn: Add support for ovn_sb_* parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/30aca97f35d22ad842a39a4625de792b1a07a246"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-16 01:20:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-09 15:06:31.000000000","tz":0},"subject":"ovn: Accept arrays for ovn_*_connection parameters","message":"ovn: Accept arrays for ovn_*_connection parameters\n\nThese parameters are regarded as StrOpt, but can accept comma-seprated\nstrings.\n\nConflicts:\n\tmanifests/provider/ovn.pp\n\nResolved conflict caused by 045272a8ce0c4e20f6ceef3bea9876fabf276524\n\nChange-Id: I725d7c167886ae799663bbe3c033fb99a81f9711\n(cherry picked from commit cf154d7172dc928b37f5453c4e7192a4b7abc6a5)\n(cherry picked from commit 2d2404649e201525d21bf39d751dda70a4f8162a)\n(cherry picked from commit 91031e6d0068fefdaa7fc3301452f2f0827b9118)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bab76fd047508841f4bddc4f2350b98d0896860d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/bab76fd047508841f4bddc4f2350b98d0896860d"}]},"branch":"refs/heads/stable/wallaby"},"3f0861d34c1ea41ac9e69cb8043357fc9f6b8485":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-11 14:55:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/849211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/11/849211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/2"}}},"commit":{"parents":[{"commit":"be42be1017cc593b907fbd8015393148e980c11a","subject":"ovn: Add support for ovn_sb_* parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/be42be1017cc593b907fbd8015393148e980c11a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-16 01:20:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 14:55:29.000000000","tz":0},"subject":"ovn: Accept arrays for ovn_*_connection parameters","message":"ovn: Accept arrays for ovn_*_connection parameters\n\nThese parameters are regarded as StrOpt, but can accept comma-seprated\nstrings.\n\nConflicts:\n\tmanifests/provider/ovn.pp\n\nResolved conflict caused by 045272a8ce0c4e20f6ceef3bea9876fabf276524\n\nChange-Id: I725d7c167886ae799663bbe3c033fb99a81f9711\n(cherry picked from commit cf154d7172dc928b37f5453c4e7192a4b7abc6a5)\n(cherry picked from commit 2d2404649e201525d21bf39d751dda70a4f8162a)\n(cherry picked from commit 91031e6d0068fefdaa7fc3301452f2f0827b9118)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f0861d34c1ea41ac9e69cb8043357fc9f6b8485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/3f0861d34c1ea41ac9e69cb8043357fc9f6b8485"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"},"2b2e58063ff6f1cbb2e470dc6ebd1a60482eaa43":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-11 18:13:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/849211/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/11/849211/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/11/849211/3"}}},"commit":{"parents":[{"commit":"59612af6f4e18d265a13ed345fd4c02ace9ea318","subject":"ovn: Add support for ovn_sb_* parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/59612af6f4e18d265a13ed345fd4c02ace9ea318"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-06-16 01:20:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 18:13:04.000000000","tz":0},"subject":"ovn: Accept arrays for ovn_*_connection parameters","message":"ovn: Accept arrays for ovn_*_connection parameters\n\nThese parameters are regarded as StrOpt, but can accept comma-seprated\nstrings.\n\nConflicts:\n\tmanifests/provider/ovn.pp\n\nResolved conflict caused by 045272a8ce0c4e20f6ceef3bea9876fabf276524\n\nChange-Id: I725d7c167886ae799663bbe3c033fb99a81f9711\n(cherry picked from commit cf154d7172dc928b37f5453c4e7192a4b7abc6a5)\n(cherry picked from commit 2d2404649e201525d21bf39d751dda70a4f8162a)\n(cherry picked from commit 91031e6d0068fefdaa7fc3301452f2f0827b9118)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b2e58063ff6f1cbb2e470dc6ebd1a60482eaa43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/2b2e58063ff6f1cbb2e470dc6ebd1a60482eaa43"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"}}]}],"submit_requirements":[]}
