)]}'
{"id":"openstack%2Fpuppet-octavia~849277","triplet_id":"openstack%2Fpuppet-octavia~master~I697fdbef762c69bb1494eae1cfa8252a2bd2c9b3","project":"openstack/puppet-octavia","branch":"master","topic":"log-offloading-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-01 09:16:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I697fdbef762c69bb1494eae1cfa8252a2bd2c9b3","subject":"Allow (more) customization of amphora log offloading","status":"MERGED","created":"2022-07-11 07:13:57.000000000","updated":"2022-08-01 09:17:28.000000000","submitted":"2022-08-01 09:16:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849277-log-offloading-opts","meta_rev_id":"b4e1dda56e579ae113d971af3d55442de1599545","_number":849277,"virtual_id_number":849277,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 09:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-07-11 22:00:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-08-01 07:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-08-01 07:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-11 09:30:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 22:00:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-08-01 07:35:32.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f43d56bdde2474baf4e9b959edd133b4da5cfb83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 07:13:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d0d23e951bcd2fdad174d27dd9b611478a94eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-11 07:15:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b49326f7c50332645e4f465de0e22c2c0724471e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:30:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab21db8fabff4cba919e699297d8038c\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81709fb682cf4fcd8306a51cae2903c3 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53dfaaacc40b41a28c7442831dd53fcd : SUCCESS in 3m 14s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d713e4f39ee549f082cfdd0e8b5b41d7 : SUCCESS in 3m 12s\n- puppet-openstack-unit-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa612d4474ee4e68bcdf813a031b8e32 : SUCCESS in 11m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f187866898c45a5b746ef1986bd6822 : SUCCESS in 8m 02s\n- puppet-openstack-unit-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae128114d43746c6ba1d6703b4e68fe8 : SUCCESS in 7m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b126ac4ed8c04852831ec38f4ffa18e3 : SUCCESS in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dbae01c1b286406c8ae439741c2faf79 : FAILURE in 20m 16s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88f8a759984f4eeeaa6aca3d29a0465a : SUCCESS in 53m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3ca72ad41a84ea496675d8dce286eb7 : SUCCESS in 17m 31s (non-voting)\n- puppet-openstack-litmus-7.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bf7085ae5c04b7c81d982843eb9c66b : SUCCESS in 24m 28s\n- puppet-openstack-litmus-7.10-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82c708d560964443892bdace46472bdc : SUCCESS in 17m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8be873b33f0b4f7abbea293a542ca8e1 : SUCCESS in 17m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec835b4399a430db6271e8134825e67 : SUCCESS in 3m 12s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fa2558d7542643108185656453e7c78e : SUCCESS in 2h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"993f34cd4895959596cd231b91af9487c4821614","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-07-11 22:00:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9bd369cb75d5e418632255a2f37771ceadb79581","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-01 07:35:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"33033ddc78c6633333f6667bf17299343b4fe530","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:37:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76a36f9ad1e64dfb4e3ee11687f4c2c0f0f11f6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:16:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50aed1a48f8c4b88b071ee48a77e4b10\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0790e39943b46c5a7882eafff96c897 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f543a427d1d4adab045054a6dd36c47 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16db2202fa934eeca7210a4a51f0973b : SUCCESS in 4m 13s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1e4190393c34a97a8187ca1d6394462 : SUCCESS in 18m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fe166beafab4d9d9791ddbe69f60d6c : SUCCESS in 12m 15s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/510438328fdc42c1b094564a08a9a8a7 : SUCCESS in 12m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/88c99b73df3f4a019da2fe077a021bcb : SUCCESS in 57m 46s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/296bf9e2b9da4405b19b88b937928f5c : SUCCESS in 35m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faff8244d98f402898ac61277d388b5c : SUCCESS in 4m 14s\n- puppet-octavia-tripleo-standalone https://zuul.opendev.org/t/openstack/build/155e62bd95cd4d30a4bad83353584664 : SUCCESS in 1h 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b21c0cb43533ba60e98044fcc3eb6550e18f9a1c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:16:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b4e1dda56e579ae113d971af3d55442de1599545","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:17:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6abcdd8b5644a7bbb99ce8f79065bd8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e3aa74b07a341ca966fe9b14815b3b8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02965a6077da2c10f936cbf3d5d262cc96bdf489","revisions":{"20f04ff45057b91e021640d9fc2d80671d186a80":{"kind":"REWORK","_number":1,"created":"2022-07-11 07:13:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/849277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/77/849277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/1"}}},"commit":{"parents":[{"commit":"76851562562f047f0efcc565317797176e1a614f","subject":"Merge \"ovn: Accept arrays for ovn_*_connection parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76851562562f047f0efcc565317797176e1a614f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 07:10:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 07:13:45.000000000","tz":540},"subject":"Allow (more) customization of amphora log offloading","message":"Allow (more) customization of amphora log offloading\n\nThis change extends coverage of [amphora_agent] options so that users\ncan customize these parameters.\n\nEspecially the log_protocol parameter is required to use TCP protocol\ninstead of UDP protocol for log transferring, for more robust logging.\n\nChange-Id: I697fdbef762c69bb1494eae1cfa8252a2bd2c9b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/20f04ff45057b91e021640d9fc2d80671d186a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/20f04ff45057b91e021640d9fc2d80671d186a80"}]},"branch":"refs/heads/master"},"02965a6077da2c10f936cbf3d5d262cc96bdf489":{"kind":"REWORK","_number":2,"created":"2022-07-11 07:15:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/849277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-octavia","ref":"refs/changes/77/849277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-octavia refs/changes/77/849277/2"}}},"commit":{"parents":[{"commit":"76851562562f047f0efcc565317797176e1a614f","subject":"Merge \"ovn: Accept arrays for ovn_*_connection parameters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/76851562562f047f0efcc565317797176e1a614f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 07:10:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-11 07:15:28.000000000","tz":540},"subject":"Allow (more) customization of amphora log offloading","message":"Allow (more) customization of amphora log offloading\n\nThis change extends coverage of [amphora_agent] options so that users\ncan customize these parameters.\n\nEspecially the log_protocol parameter is required to use TCP protocol\ninstead of UDP protocol for log transferring, for more robust logging.\n\nChange-Id: I697fdbef762c69bb1494eae1cfa8252a2bd2c9b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/02965a6077da2c10f936cbf3d5d262cc96bdf489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-octavia/commit/02965a6077da2c10f936cbf3d5d262cc96bdf489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
